)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~749907","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I48edcc48549e811aa96091f8ce838659d250ae04","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"I48edcc48549e811aa96091f8ce838659d250ae04","subject":"Fix native service path","status":"MERGED","created":"2020-09-04 09:12:23.000000000","updated":"2020-09-07 08:57:47.000000000","submitted":"2020-09-07 08:51:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749907-1599468702817-9d651cf1","meta_rev_id":"4e657bc7f73adb682322bc4ad999faa9b612551f","_number":749907,"virtual_id_number":749907,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 08:51:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-09-07 07:24:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-06 18:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-09-07 07:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-06 18:30:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-07 07:24:38.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-09-07 08:51:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d6b88adf67fabd7b67e4803900cbbb2fecf17c1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-04 09:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ddc6247a0ea64af3ddaf67c7c46f9cd39b1fa89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 11:36:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/75e8fc8afd584abcb384d37b15468b0c : SUCCESS in 8m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9338852f2521458bb9889446b45b86fc : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cdb069066c3344059b8bb7f2063d5e26 : FAILURE in 1h 06m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8deacecf779b454b9745ae4787f9651f : SUCCESS in 1h 12m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b765723cc468482fa0ad883feedd5a65 : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68f8c3d31c774f6e9ae37a918f3e85ef : SUCCESS in 1h 05m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/38b186ce147b43479d7d219ee612123b : FAILURE in 2h 13m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0f072c6ae424e4898410ef67da3d099 : SUCCESS in 1h 53m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9ff28df1772b4524994c04cf30bce866 : SUCCESS in 1h 02m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a85ba77d4164b07975849309d2d2d66 : SUCCESS in 1h 01m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/678f178f6c5a4393842f3036b606149e : SUCCESS in 1h 02m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2956a3b7a31441698d8f8221d67112b : SUCCESS in 55m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49184af32984a288e46f6f6710de0a7 : SUCCESS in 5m 36s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/90707d9b061f43fb9d94753b739a39da : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":1},{"id":"aeaaf119ebd4f9cc7e20e4675bcc272503519e43","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-06 13:00:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f178a94d0e8d11b59f42e8f7fdc868559f3f5b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 15:13:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/84283aa5f39443098a7286d3d6af5c49 : SUCCESS in 5m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a1c50794d4ce48dd86d1829d0379a8fe : SUCCESS in 1h 03m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/70aa64cd700d40ee8609b9ad0c86d4d3 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/935b6a76d65d4a459efe4c7eaf7965da : SUCCESS in 1h 06m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09b0ccdac0814d7993fa1a6a36b835d2 : SUCCESS in 57m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cbd821075afc4bfd9194dd36875dd82e : SUCCESS in 2h 03m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8dd33629f784980b13aab618a4c6d4d : SUCCESS in 2h 00m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/23a4191d246a488a9213963ffa68ca90 : SUCCESS in 55m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27bae4be075a46118dbf6d4f525080c1 : SUCCESS in 1h 21m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9383d11bb674b70becbe9ca92d42b88 : SUCCESS in 4m 26s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b41ab3554fc84d82ab506ecfefacb38f : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6ac14d250a422de21a7e83d4ca326373496cd8f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-06 18:30:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05bbf6ebb8e1ba63ba7d1095aebd67f9b05ca70f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-07 07:24:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95cf474dd5c4eb7cec66614a8029b0080212647d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 07:24:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3da72bc2808c6a46aba2ecf1791a4a1941100745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:51:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d4a4a8db356998f62cce72fa97a8914bc39fd6ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:51:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5ebd1bb973904cfda04f6cd2818e8bce : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e3d4f357f9f94920a965682e8fc6ed3a : SUCCESS in 1h 06m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7b6c654f4625427496be7619e443f530 : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/159d8a4114de48fabfe1dbc5eb7a724a : SUCCESS in 1h 10m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2ed150690d84d36af47b148ed92d3c4 : SUCCESS in 1h 04m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fcbb19b35d4742f2a9f340811f7f9d40 : SUCCESS in 57m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02f11013d55c455c837c44523d135ed9 : SUCCESS in 54m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82431e4c5bbe4987af113fa48855f0d0 : SUCCESS in 7m 15s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/cdf7bb0b6b0d46c4b76292924abfdf4f : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"77e04a196524f70b1a3e4c041f294a095ef77752","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e920e59cf244a0e93b8b4b9439a592b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1},{"id":"4e657bc7f73adb682322bc4ad999faa9b612551f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-07 08:57:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit b09b92d6658076552222e16987a0a2dd0ba083f5","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"512f5736ebdf917f4756c998acf028ac0b80ee8b","revisions":{"512f5736ebdf917f4756c998acf028ac0b80ee8b":{"kind":"REWORK","_number":1,"created":"2020-09-04 09:12:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/07/749907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/07/749907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/749907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/749907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/749907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/07/749907/1"}}},"commit":{"parents":[{"commit":"ab5de45fa74f23d58bebdda75e5232da946f3d18","subject":"Merge \"Use the utility host for db setup tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ab5de45fa74f23d58bebdda75e5232da946f3d18"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 09:12:12.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-04 09:12:12.000000000","tz":180},"subject":"Fix native service path","message":"Fix native service path\n\nWhen glance_use_uwsgi is set to false, glance is launched with native\nservice instead of the uwsgi. For that scenario, execstarts key is used\nwhich had wrongly defaulted config-file location.\nEventually, providing this option is not required and we may rely on defaults\n\nChange-Id: I48edcc48549e811aa96091f8ce838659d250ae04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/512f5736ebdf917f4756c998acf028ac0b80ee8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/512f5736ebdf917f4756c998acf028ac0b80ee8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
