)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~751270","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ida1d88be3ae8f52b6c467563a9b99a7f7d44c4c0","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"run_condition","hashtags":[],"change_id":"Ida1d88be3ae8f52b6c467563a9b99a7f7d44c4c0","subject":"Define condition for the first play host one time","status":"MERGED","created":"2020-09-11 12:50:11.000000000","updated":"2020-09-21 08:42:16.000000000","submitted":"2020-09-21 08:38:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751270-1600677483446-2f639b6f","meta_rev_id":"d75b62c35683f47098861db15db336484b75576b","_number":751270,"virtual_id_number":751270,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 08:38:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-09-21 07:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 06:43:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-09-21 07:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 06:43:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-21 07:09:53.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-09-21 08:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9329a61124e46cd77198dd5055dad87aa89a73a4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-11 12:50:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a71be99a67337bde5076894b064ae4b45287228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:24:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a1a99955cb304cef8153c6310ed77ef4 : SUCCESS in 8m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4965af1d1e2c407e8d2872cc9fd4deab : SUCCESS in 1h 06m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/18ef05bdbd0f44eda72af04c523797fd : SUCCESS in 1h 53m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dae23f43511b420dadb78f048c4db683 : SUCCESS in 1h 10m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecc8eec5e4f1481eb002359bb6571e41 : SUCCESS in 1h 01m 09s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6aa5b9c1d93f4cfba01ee43ccf1154f3 : SUCCESS in 2h 11m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6141f007ed854c0f9859ed98431f3a5c : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/774064929f8d418a81cac9fccac9c2eb : SUCCESS in 56m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fef74aaa95df4174af595b255b115f77 : SUCCESS in 1h 08m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a351a73be6459abdea3154214c87c1 : SUCCESS in 13m 56s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/3be5a888bb334bf5aaeac479522ce4a6 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e50ece3f651bc489675d0d990aac8640e1053a66","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-11 19:14:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99b34843d695c9811402876ae9a8a0e523b50efa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-15 15:59:32.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a36ccbbb81e488dae31970216d0f8a9459ce86c0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 11:54:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d7f6410338953ad32fab47e46b0ae8753fe63e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:05:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1de0304b0e94f62958f2dc7fbde6224 : SUCCESS in 7m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a65dd4c84b4047f79b801db9532249b5 : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/838a62286aaa42ad87955c8c09894ab4 : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d28026852ff428d8ed5f6ec2ac328b5 : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c18da0412d5d46949275c4c37c17f9a0 : SUCCESS in 1h 31m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/95f913ef32b6457299a92c36fbbe8d9f : SUCCESS in 2h 01m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adc0a4f0d06f4d9cb2c147dcb279f7d7 : SUCCESS in 1h 40m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c83c268ee0d04d1da0f7a36c29db8b99 : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6299935aaca64b65a2b47e88e2cb951c : SUCCESS in 55m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06fdfd55c5042f7a83ca0c526ff62bf : SUCCESS in 25m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/f242fb5e772c4b6582173530c681fc06 : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e0781d7eaabaceeeeed28f6bdbae30991b8981b1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-18 06:43:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2f89da88ace61ff2bc182d2ae8a671fbac9d939d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-21 07:09:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0eabe628ba1118d998b14f10d2728ff4750aaf68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:10:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf6989ad3a64663d768c03b67974abb359b31fca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:38:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c8aa32e49f7741cdac0a9bbdcfdf96a8 : SUCCESS in 7m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d29cd1a06c1c474d9752954b622c8d11 : SUCCESS in 59m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fef92656365d46e7a1f6696126cd765e : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dccb3fd8a3c4597843817894d743ef8 : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af0281077ef14ef8a18413d29f323f77 : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/04fa8ba004524541a493e4e4dca4332e : SUCCESS in 59m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a16861ebda8499c9a84f6a972d49d48 : SUCCESS in 52m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c93cd37939425e977f12987ba51da5 : SUCCESS in 14m 59s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9b34bca48ca24a23970ad40d5d095052 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7fd7f97508e021c5b23fe5b40f3d9bc148e69d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:38:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a66e66f63b2cd9904bf191c29869cfaac2d6dda4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:42:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/146c16b1e94443d5a0d12aec713ddbc5 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c2f04b0283dbf0482285036eeea3bc8f7a14e19","revisions":{"74866bd23b4c1981b403af4a51a6130e1a039c2a":{"kind":"REWORK","_number":1,"created":"2020-09-11 12:50:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/751270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/70/751270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/1"}}},"commit":{"parents":[{"commit":"ab5de45fa74f23d58bebdda75e5232da946f3d18","subject":"Merge \"Use the utility host for db setup tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ab5de45fa74f23d58bebdda75e5232da946f3d18"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:49:52.000000000","tz":180},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Ida1d88be3ae8f52b6c467563a9b99a7f7d44c4c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/74866bd23b4c1981b403af4a51a6130e1a039c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/74866bd23b4c1981b403af4a51a6130e1a039c2a"}]},"branch":"refs/heads/master"},"1c2f04b0283dbf0482285036eeea3bc8f7a14e19":{"kind":"REWORK","_number":2,"created":"2020-09-16 11:54:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/751270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/70/751270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/751270/2"}}},"commit":{"parents":[{"commit":"ab5de45fa74f23d58bebdda75e5232da946f3d18","subject":"Merge \"Use the utility host for db setup tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ab5de45fa74f23d58bebdda75e5232da946f3d18"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:49:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-16 11:54:27.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Ida1d88be3ae8f52b6c467563a9b99a7f7d44c4c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/1c2f04b0283dbf0482285036eeea3bc8f7a14e19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/1c2f04b0283dbf0482285036eeea3bc8f7a14e19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
