)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~756681","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I4f230347a9c464244a6bac42e4234b627ca460ed","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"uwsgi-restart","hashtags":[],"change_id":"I4f230347a9c464244a6bac42e4234b627ca460ed","subject":"Trigger uwsgi restart","status":"MERGED","created":"2020-10-08 11:53:33.000000000","updated":"2020-10-12 09:30:54.000000000","submitted":"2020-10-12 09:28:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756681-1602494937241-75899b5c","meta_rev_id":"c6087466a260c231adcf78d363d2b04ca7e8e8bb","_number":756681,"virtual_id_number":756681,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 09:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-12 07:27:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-09 19:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-12 07:27:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 19:52:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-12 07:27:01.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-12 09:28:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57f2d2982107bcb9d42786cbc1f893858d20c9d0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 11:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36a3feb892a68af53f49d2a9da5623d7239aceed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-08 12:00:14.000000000","message":"Topic set to uwsgi-restart","accounts_in_message":[],"_revision_number":1},{"id":"a82ca28823f9e17eedd53f888c771604100a5aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:05:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1c38e7c56d424500b0df8c6f87800605 : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f9c71dda45c940bd833ee86302047e61 : FAILURE in 1h 07m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/41c09e63565b403393bd7152817b0526 : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/129ea08289a741b99276f799094aede2 : SUCCESS in 1h 11m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/680ebc7acac941c89e9429891a38f72b : SUCCESS in 1h 00m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/674fccaf0f7a4d0c9b3a02e9c0459e0d : SUCCESS in 2h 01m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d5c6861de474972803eab88bc5f8068 : SUCCESS in 1h 36m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/21b31d54f7d846768b8ff90ea2b09081 : SUCCESS in 1h 02m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f691ad91161d4f72b17223e190e51b12 : SUCCESS in 54m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ee5f3232a845e791f7655c3e15bca4 : SUCCESS in 5m 06s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4af5f2d461d44da0830f1647b4872938 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":1},{"id":"28180b0925e4f6d65ccfbc5bb56654108ad78ab6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-08 19:12:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0aac8b3fb252a896d7dd84fa2fd66f85903f2144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:40:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2a3d856ce63a4058878102ecf49b4019 : SUCCESS in 8m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bcdecca57bb041208453cfc3155c6b48 : SUCCESS in 1h 05m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a859dc748e364bfb8cea26c0d0d8b81e : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1544d62b636f4985bb72220a66507d0c : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33c47a9d759742d2b19042df246c6901 : SUCCESS in 1h 11m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b61e9b6bb3f54dc4b8ad96c323d831b3 : SUCCESS in 2h 09m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21fce6f3c50549fc9a01003bb5b92521 : SUCCESS in 1h 47m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fb198a176f234eb0af21173580108e3e : SUCCESS in 1h 02m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d76d86c81dfb470a90bd1c3193550fda : SUCCESS in 55m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b53f096a6945ce9abdc3ec64cbc17c : SUCCESS in 5m 42s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b55f187161754cb0be19f646be3b43b9 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8075fdcb5297bbbfde6fa8e2caf5bcea341eec7c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-09 19:52:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8ea05e09300b9e8780cfa219effae750f09a56f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-12 07:27:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3d3c72c8acd25e3f4016c539aaedfec1693333d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 07:27:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d1a4339e7481e87d58c2cddb4fe2fd2ad738dff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:28:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/24f8a6fec16a471f8a8c742aba30e45f : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7d43dd13a1d341eab7a02ec269b783a6 : SUCCESS in 1h 03m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ff300f71e37c4baf83df196cff2dd09b : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49701fc635ac4069ad69f299a89a1514 : SUCCESS in 1h 47m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29f41e6b79014ae5986a9a5fdae7462b : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e490203e467f444f96e0090c9a0398e9 : SUCCESS in 1h 08m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db591362a5f4462091dd28ac1c99653e : SUCCESS in 1h 04m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94d51119cc1942858714333a92cbcc08 : SUCCESS in 5m 51s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/8c8af4a6c5664a85b0970634ac121711 : SUCCESS in 43m 17s","accounts_in_message":[],"_revision_number":1},{"id":"19df1991f3f0cd7df8973208f6ade00166888121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:28:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6087466a260c231adcf78d363d2b04ca7e8e8bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 09:30:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2568d26ae44175b22b250f8bfc7b5a : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a51937db34ac17a2426263a983569e1bc81c9125","revisions":{"a51937db34ac17a2426263a983569e1bc81c9125":{"kind":"REWORK","_number":1,"created":"2020-10-08 11:53:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/756681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/81/756681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/756681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/756681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/756681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/756681/1"}}},"commit":{"parents":[{"commit":"4a869ef6b86c30600b4abb1ef9f7dfe5be988876","subject":"Merge \"Reduce number of processes on small systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4a869ef6b86c30600b4abb1ef9f7dfe5be988876"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-08 11:53:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-08 11:53:21.000000000","tz":180},"subject":"Trigger uwsgi restart","message":"Trigger uwsgi restart\n\nWhen we were migrating service to uwsgi usage, we clean forgot to\ntrigger uwsgi restart on service config change.\n\nChange-Id: I4f230347a9c464244a6bac42e4234b627ca460ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a51937db34ac17a2426263a983569e1bc81c9125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a51937db34ac17a2426263a983569e1bc81c9125"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
