)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~760581","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I775d882382b06110358879034f735a8c6e73ec93","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"ceph_client_distro","hashtags":[],"change_id":"I775d882382b06110358879034f735a8c6e73ec93","subject":"Do not symlink ceph libraries for distro path","status":"MERGED","created":"2020-10-30 15:33:06.000000000","updated":"2020-11-02 10:39:13.000000000","submitted":"2020-11-02 10:19:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760581-1604312387222-846bd11f","meta_rev_id":"259c9008a85265b84f10c38ac09a9444a719d65e","_number":760581,"virtual_id_number":760581,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-02 10:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-11-02 08:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-10-30 17:32:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-01 17:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-11-02 08:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 17:32:46.000000000","updated_by":{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},"reviewer":{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},"state":"REVIEWER"},{"updated":"2020-11-01 17:15:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-02 08:43:19.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-11-02 10:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7333b6f89eb7af1b4c46c2569881c079daf6afde","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-30 15:33:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4160d0c923ce046b4e04241b1c74e7b51ca6043","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-30 15:33:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5eda388e9af9162e56910aaf95862e11388d8f41","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-30 15:36:00.000000000","message":"Topic set to ceph_client_distro","accounts_in_message":[],"_revision_number":2},{"id":"cb675ca825285b88471ea2a7c221a282e8c61e40","author":{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},"date":"2020-10-30 17:32:46.000000000","message":"Patch Set 2: Code-Review+1\n\ntested on our dev Ubuntu Train it works.","accounts_in_message":[],"_revision_number":2},{"id":"757953bb58314428cfc3a20d9c11ab5726ded50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 17:46:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a64b1fca8cf242d292b1312960da644b : SUCCESS in 14m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/16754afc2e5e491e8e99772af3cf57a1 : SUCCESS in 1h 05m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/241e5ea3041447abbd0bdbdd30ecde17 : SUCCESS in 1h 33m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f7043b2dde34c25854831a60b43975c : SUCCESS in 1h 24m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cb1705a353542fb93181c46345560e5 : SUCCESS in 1h 10m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f2173fe0579d49eea32bf24473478cb6 : SUCCESS in 2h 05m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edc9c04cb82d42529737bc5ab030df4e : SUCCESS in 1h 30m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/160dd315478e4496bb77bfc45d582efd : SUCCESS in 1h 17m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afe7bfc888694cccb26ea36c5bef84bc : SUCCESS in 54m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e5084b54434067909d6c753bc63aea : SUCCESS in 5m 23s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/5bb3a99a131b40a4b42c1ed80c0e51ff : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4ef1dd808e36a844f7162734800b651283364ed7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-01 17:15:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cb06e83b3889c01f2324f94cc5c51c303205c2af","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-11-02 08:43:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"776dc478beb198084840891effa03f52d9501a6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 08:43:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c484104605f300031a2373c4b1ba4a6daf0fa7b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:19:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d7f88338c4c24e74bc0a2a119c2c505e : SUCCESS in 9m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/403b8e7ce28941c8ac401ce5e3a504f1 : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/576e341c55a8435eadcff8c444143dac : SUCCESS in 1h 19m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e8b3493ed7e4962be118044d7b850c3 : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4744e600728b40b38b7e622d196594f7 : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dcaa0c57e450430f8e0ee6dc93af1e27 : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aefdbc3986d943c8a4bb06806c12f010 : SUCCESS in 57m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad4013188264bcc9b5a366ab76c8bf8 : SUCCESS in 6m 13s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9e60207f1db042dbba51cb1a0ce78072 : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":2},{"id":"971d2ec04b83711f9444e3e0f5c76e3de9dabe9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:19:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fecefdbbf1d3e0aecae9e4c31b491550459759ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:22:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2a03f0718245eab85ff2c5051493da : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f9ffeb3b590c7e2ea060633db06d7d8cb888318b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-02 10:38:30.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 36f2240d1b47fd54c20af6af3ae703c743911dab","accounts_in_message":[],"_revision_number":2},{"id":"8343cf4f84b03f3550949b3158ad902065120c3a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-02 10:38:54.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d948fb30af4ebca9c3c8ce1ed979bcfb78f6dac9","accounts_in_message":[],"_revision_number":2},{"id":"259c9008a85265b84f10c38ac09a9444a719d65e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-02 10:39:13.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 52483e1dcae326144c467fbc3dfef280a0886644","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3876d573648b21adee9d49ce2bc27b84846ea47d","revisions":{"fe7e3c8e5bc8df7f1ae85ce656f3b904e346c8cf":{"kind":"REWORK","_number":1,"created":"2020-10-30 15:33:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/760581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/81/760581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/1"}}},"commit":{"parents":[{"commit":"1db1e5a3679507900e013740dfee55783d38be27","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/1db1e5a3679507900e013740dfee55783d38be27"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-30 15:32:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-30 15:32:54.000000000","tz":120},"subject":"Do not symlink ceph libraries for distro path","message":"Do not symlink ceph libraries for distro path\n\nceph_client role checks length of the `openstack_service_venv_bin` variable\nto determine if libraries symlinking into venv should occur. So for\ndistro path this should be empty string as no symlinking should be done.\n\nChange-Id: I775d882382b06110358879034f735a8c6e73ec93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fe7e3c8e5bc8df7f1ae85ce656f3b904e346c8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fe7e3c8e5bc8df7f1ae85ce656f3b904e346c8cf"}]},"branch":"refs/heads/master"},"3876d573648b21adee9d49ce2bc27b84846ea47d":{"kind":"REWORK","_number":2,"created":"2020-10-30 15:33:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/760581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/81/760581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/81/760581/2"}}},"commit":{"parents":[{"commit":"1db1e5a3679507900e013740dfee55783d38be27","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/1db1e5a3679507900e013740dfee55783d38be27"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-30 15:32:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-30 15:33:44.000000000","tz":120},"subject":"Do not symlink ceph libraries for distro path","message":"Do not symlink ceph libraries for distro path\n\nceph_client role checks length of the `openstack_service_venv_bin` variable\nto determine if libraries symlinking into venv should occur. So for\ndistro path this should be empty string as no symlinking should be done.\n\nChange-Id: I775d882382b06110358879034f735a8c6e73ec93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3876d573648b21adee9d49ce2bc27b84846ea47d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3876d573648b21adee9d49ce2bc27b84846ea47d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
