)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~764169","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ide3602cf78745580226001d25817e791f159de5a","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"Ide3602cf78745580226001d25817e791f159de5a","subject":"Dep\u0027s should be restricted by tox-constraints","status":"ABANDONED","created":"2020-11-25 13:53:41.000000000","updated":"2020-11-27 01:07:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f9330031e10ebaf7a5aaa444306318fd76ee4b9","_number":764169,"virtual_id_number":764169,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-25 14:45:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"all":[{"value":-1,"date":"2020-11-26 13:46:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 14:45:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 13:46:01.000000000","updated_by":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"reviewer":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"state":"REVIEWER"}],"messages":[{"id":"292b9b6f5d68fd28586cbe842532298e76ff0371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-25 13:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e6086f377d52111aad1943311899e6b2ec0e2d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:45:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/523218ca1c6a4da78d232eada75dfec5 : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d1a62c66234718991079e178e11d5d : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d74fe06a800340c2bf9840df74071e40 : SUCCESS in 3m 26s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/229184d8f92f49719fdf9fdf1dda366d : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4fa0ea4457a7ac8940de0948bb9df91a110e8f8b","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-11-26 13:46:01.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"7f9330031e10ebaf7a5aaa444306318fd76ee4b9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-27 01:07:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd2d599ee4374d2a0764be43fb57d58f18c984ef","revisions":{"cd2d599ee4374d2a0764be43fb57d58f18c984ef":{"kind":"REWORK","_number":1,"created":"2020-11-25 13:53:41.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/764169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/764169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/764169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/764169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/764169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/764169/1"}}},"commit":{"parents":[{"commit":"7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-24 15:34:46.000000000","tz":0},"committer":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-25 13:51:55.000000000","tz":0},"subject":"Dep\u0027s should be restricted by tox-constraints","message":"Dep\u0027s should be restricted by tox-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s tox-constraints.txt\n\nChange-Id: Ide3602cf78745580226001d25817e791f159de5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cd2d599ee4374d2a0764be43fb57d58f18c984ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cd2d599ee4374d2a0764be43fb57d58f18c984ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
