)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~768630","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I0d6d392a90c46123d30271ce32993664009395d3","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"I0d6d392a90c46123d30271ce32993664009395d3","subject":"Add Victoria release info","status":"ABANDONED","created":"2020-12-29 02:23:09.000000000","updated":"2020-12-29 06:32:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6284ce72303a5c07c88bbd65648132d730b2f1ac","_number":768630,"virtual_id_number":768630,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-29 03:53:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 03:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"163a809e68e384d0c6ae27242b2ee84af485af42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2020-12-29 02:23:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3185b1713b9d2e28a31d4fb1c61cfc9c19ba64e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 03:53:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/392adc295a604e8dbe6ae16b5c7430ff : SUCCESS in 7m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7d20616408064e06a0f7faee51eaddea : SUCCESS in 1h 02m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f52adc9368954fb9a6468dbb028d429a : SUCCESS in 1h 04m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cec5caaa7c244a4b941add02014d154c : SUCCESS in 1h 02m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ac128aa0ada4bccbf8ac1bd96c3807e : SUCCESS in 1h 00m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e3a89dca68a4084881c3a420ffb7323 : SUCCESS in 1h 25m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ae66c312a34b91b6e5358a55d02a50 : FAILURE in 48m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a71763a68e354ab5ae078b29fc697b81 : SUCCESS in 49m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0b6bea32e134d819b0f3fd1087d2445 : SUCCESS in 50m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c053fbbd62349ac97187f59db62e36d : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bee7cefd099474381d1b3281a3ab689 : SUCCESS in 2m 57s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d7ca5d4c281045f18a6dfef03203b418 : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6284ce72303a5c07c88bbd65648132d730b2f1ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2020-12-29 06:32:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15cc9e4f92530bcf1ea67d2c6ebce57aab8fe41c","revisions":{"15cc9e4f92530bcf1ea67d2c6ebce57aab8fe41c":{"kind":"REWORK","_number":1,"created":"2020-12-29 02:23:09.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/30/768630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/30/768630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/768630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/768630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/768630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/30/768630/1"}}},"commit":{"parents":[{"commit":"7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 02:21:28.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 02:23:02.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I0d6d392a90c46123d30271ce32993664009395d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/15cc9e4f92530bcf1ea67d2c6ebce57aab8fe41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/15cc9e4f92530bcf1ea67d2c6ebce57aab8fe41c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
