)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~768678","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I7311bde86d641e5e6495c7069c705783e2a5ee87","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"I7311bde86d641e5e6495c7069c705783e2a5ee87","subject":"Add Victoria release info","status":"ABANDONED","created":"2020-12-29 06:36:40.000000000","updated":"2021-01-26 17:24:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7e94109f04e410b7c0af8f93b59af2083673dc25","_number":768678,"virtual_id_number":768678,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-30 01:13:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 06:49:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8774823458b3ceec4a8ba1af3609623253494c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2020-12-29 06:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77043cd9ca32e034201cab7dbf2177063b576a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 06:49:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fa0df16679f248f18f2e59aba97295ad : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccec0dc7df5742f3bd974ca56869ff10 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0463a1fb5714d23864a1277f1c8980d : FAILURE in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"57b06db40a128b49ac58656cd180138247ffbcea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2020-12-30 01:02:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e901b6de645dca9d2133ee2e1a40f8465b257b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-30 01:13:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4a12a0e94c0d4eeca324aa9218f3de08 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d43ef44bd84d64b99d511cc1eef248 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a23a0de6cb5840b09598f3781054fc3a : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"7e94109f04e410b7c0af8f93b59af2083673dc25","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 17:24:22.000000000","message":"Abandoned\n\nWe are no longer publishing releasenotes for this repo","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5529dc7d18e962e7c9244838b97cc23030b8fb5","revisions":{"660d1a9f35ba189eecf727ba2c9a5f734427fe2a":{"kind":"REWORK","_number":1,"created":"2020-12-29 06:36:40.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/78/768678/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/78/768678/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/1"}}},"commit":{"parents":[{"commit":"7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 06:35:35.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 06:36:35.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I7311bde86d641e5e6495c7069c705783e2a5ee87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/660d1a9f35ba189eecf727ba2c9a5f734427fe2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/660d1a9f35ba189eecf727ba2c9a5f734427fe2a"}]},"branch":"refs/heads/master"},"e5529dc7d18e962e7c9244838b97cc23030b8fb5":{"kind":"REWORK","_number":2,"created":"2020-12-30 01:02:26.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/78/768678/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/78/768678/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/78/768678/2"}}},"commit":{"parents":[{"commit":"7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7c9b4258fa6f9288fff25b219e88cb0c9ddfb3c0"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 06:35:35.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-30 01:01:55.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I7311bde86d641e5e6495c7069c705783e2a5ee87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e5529dc7d18e962e7c9244838b97cc23030b8fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e5529dc7d18e962e7c9244838b97cc23030b8fb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
