)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~772850","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Id9a667721909f7111da85a9a5b088a661ad19c73","project":"openstack/openstack-ansible-os_glance","branch":"master","hashtags":[],"change_id":"Id9a667721909f7111da85a9a5b088a661ad19c73","subject":"Added option to congfigure an image upload larger than the 1Tb default upload size.","status":"ABANDONED","created":"2021-01-28 10:43:57.000000000","updated":"2021-01-28 13:27:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c33527ecba405e1384ec77a0ef342f1ee15bb599","_number":772850,"virtual_id_number":772850,"owner":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"date":"2021-01-28 13:26:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 13:15:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b2b5f11b8a6dead89740f2529832a3d2009d164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"date":"2021-01-28 10:43:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bd8376ad1d579ae95d004b2ea77e08679b5e207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:15:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/891eec62c02742ee870b47104abb6e83 : SUCCESS in 7m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/549fa0eb52e8404aa39308d03b3c2c4a : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/55f1defcd5844b5880e60582c5d5ed69 : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2571b04cc2a649e9ae0f52f1d3e84b13 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2d7341926e74a55aee9814b23b39a57 : SUCCESS in 1h 19m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b42c59219a0147bdbdef0fc36d5cb237 : SUCCESS in 2h 02m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51056c105e3342a0b020f0a03eb2d6d0 : SUCCESS in 1h 43m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/81df9fc5f2b748a8a66ecf113dffae31 : SUCCESS in 2h 22m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7e01045391f473fafaf5ef0fe64979e : SUCCESS in 56m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2348011e613471f9fb394a6379a67de : SUCCESS in 5m 38s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4ab5e1e1dcea4479bea8f18705f36734 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ee79dfa133335be36da343b5e9d96e6ddcea8d40","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"date":"2021-01-28 13:26:58.000000000","message":"Removed Verified+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":1},{"id":"c33527ecba405e1384ec77a0ef342f1ee15bb599","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"date":"2021-01-28 13:27:12.000000000","message":"Abandoned\n\nNot relevant","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c3b9a4d1386a2fb42752b704d1485fa753cd86e","revisions":{"3c3b9a4d1386a2fb42752b704d1485fa753cd86e":{"kind":"REWORK","_number":1,"created":"2021-01-28 10:43:57.000000000","uploader":{"_account_id":27915,"name":"PerToft","email":"per@minfejl.dk","username":"PTO"},"ref":"refs/changes/50/772850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/50/772850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/772850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/772850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/772850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/772850/1"}}},"commit":{"parents":[{"commit":"ab46154a2557dc5812a3c6613a0eebcb6df21cc3","subject":"[reno] Stop publishing release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ab46154a2557dc5812a3c6613a0eebcb6df21cc3"}]}],"author":{"name":"Per Abildgaard Toft","email":"pto@netic.dk","date":"2021-01-28 10:40:38.000000000","tz":60},"committer":{"name":"Per Abildgaard Toft","email":"per@minfejl.dk","date":"2021-01-28 10:43:40.000000000","tz":60},"subject":"Added option to congfigure an image upload larger than the 1Tb default upload size.","message":"Added option to congfigure an image upload larger than the 1Tb default upload size.\n\nCloses-Bug: 1913563\nChange-Id: Id9a667721909f7111da85a9a5b088a661ad19c73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3c3b9a4d1386a2fb42752b704d1485fa753cd86e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3c3b9a4d1386a2fb42752b704d1485fa753cd86e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
