)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~776651","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ie727d3a83b12388f36178d00eb5543f0733e077c","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/glance","hashtags":[],"change_id":"Ie727d3a83b12388f36178d00eb5543f0733e077c","subject":"Remove deprecated option show_multiple_locations","status":"NEW","created":"2021-02-19 14:41:23.000000000","updated":"2021-02-19 16:29:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b48942300c2c506c4de3a8204fd309cddc5d58f8","_number":776651,"virtual_id_number":776651,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-19 16:29:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-1,"date":"2021-02-19 15:28:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 15:27:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-02-19 15:28:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-19 16:29:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb94d76f0ff93f715d81903ed76c65bc114b34b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-02-19 14:41:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65642b648d261418e36572092a217102ade3eadf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-02-19 14:47:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83b62db051f1fa1616b925f978ca46c9c42fa56a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-19 15:27:42.000000000","message":"Patch Set 2:\n\nWe can not drop this option because while it\u0027s deprecated, it functionality can\u0027t be worked around:\n`You cannot work around this option via policy configuration at the present time, though that is the direction we believe the fix will take.`\n\nAlso ceph backend never contains credentials, and we enable it only for rbd backend. Also deployer can override it in user_variables.yml","accounts_in_message":[],"_revision_number":2},{"id":"cd71771999f2124c830436dec9e6d11f6321f4ff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-19 15:28:00.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"b48942300c2c506c4de3a8204fd309cddc5d58f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:29:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4799564ceed5471888a6a00c9f2540de : SUCCESS in 9m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/02bf1dc855374fcfbfa60c6c825c0930 : SUCCESS in 1h 09m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6b0e1b0704024c39a4749abb07f81c80 : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7aaa6ff47db641e38f83037a25a4d9ce : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d18f9cd1e13b4995b0766d940a85b5cc : SUCCESS in 1h 02m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2e78b6c970c54aca9e89d2936cbd2311 : SUCCESS in 1h 21m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f036d2393e854d53aa073cf328fc5da7 : FAILURE in 1h 36m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f44e08faa3054c16ba367e49de5d1286 : SUCCESS in 59m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ba6852efda84f31a0b2c06cc21d6492 : SUCCESS in 51m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787aaa4edf4a41a580bdbb9f777b96d2 : SUCCESS in 5m 45s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/d1e0b5a868df4bd081332a1746752926 : FAILURE in 31m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"939bfebdd090746fff3c28172d4cda2f7cbed5d8","revisions":{"35792565d725a9adf0980e92a3c11b237f6064a2":{"kind":"REWORK","_number":1,"created":"2021-02-19 14:41:23.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/51/776651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/51/776651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/1"}}},"commit":{"parents":[{"commit":"b9623c530b0c5dc7019e9feeaf480cb9142cf8c0","subject":"Merge \"Cleanup glance variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b9623c530b0c5dc7019e9feeaf480cb9142cf8c0"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-19 14:36:56.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-19 14:36:56.000000000","tz":0},"subject":"Remove deprecated option show_multiple_locations","message":"Remove deprecated option show_multiple_locations\n\nRevealing image locations can present a GRAVE SECURITY RISK as image locations can sometimes include credentials.\nhttps://docs.openstack.org/glance/victoria/configuration/glance_api.html#DEFAULT.show_multiple_locations\n\nChange-Id: Ie727d3a83b12388f36178d00eb5543f0733e077c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/35792565d725a9adf0980e92a3c11b237f6064a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/35792565d725a9adf0980e92a3c11b237f6064a2"}]},"branch":"refs/heads/master"},"939bfebdd090746fff3c28172d4cda2f7cbed5d8":{"kind":"REWORK","_number":2,"created":"2021-02-19 14:47:01.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/51/776651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/51/776651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/51/776651/2"}}},"commit":{"parents":[{"commit":"b9623c530b0c5dc7019e9feeaf480cb9142cf8c0","subject":"Merge \"Cleanup glance variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b9623c530b0c5dc7019e9feeaf480cb9142cf8c0"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-19 14:36:56.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-02-19 14:46:54.000000000","tz":0},"subject":"Remove deprecated option show_multiple_locations","message":"Remove deprecated option show_multiple_locations\n\nRevealing image locations can present a GRAVE SECURITY RISK as image locations can sometimes include credentials.\nhttps://docs.openstack.org/glance/victoria/configuration/glance_api.html#DEFAULT.show_multiple_locations\n\nChange-Id: Ie727d3a83b12388f36178d00eb5543f0733e077c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/939bfebdd090746fff3c28172d4cda2f7cbed5d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/939bfebdd090746fff3c28172d4cda2f7cbed5d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
