)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~780722","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I7eff6aa9bc0e3d635946343b2602ec9431794de3","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa-facts","hashtags":[],"change_id":"I7eff6aa9bc0e3d635946343b2602ec9431794de3","subject":"Convert remaining ansible_ fact variable to ansible_facts[]","status":"MERGED","created":"2021-03-16 08:03:30.000000000","updated":"2021-03-17 13:49:39.000000000","submitted":"2021-03-17 13:48:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780722-osa-facts","meta_rev_id":"2e0658b85a3d7edee08bdebfa1aa4291b83f270f","_number":780722,"virtual_id_number":780722,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-17 13:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-03-16 15:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-03-16 14:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-03-16 15:17:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 10:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 14:10:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-03-16 15:17:20.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"4e44b6c25d1b861f093f13291549b9c7bb84b838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 08:03:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47a0ede87cb907105e2d8b8df883d3d93de5d3f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 10:04:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9b70288936c04e87972dc65e1004ab6d : SUCCESS in 10m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ca9b9083f3bc44e2be3474da59b87d2d : SUCCESS in 1h 16m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bed9afa5a0f843af8ad320b4454b267b : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d91d2b7567c24199b82c4dce5c6c5ad8 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b82c20a2dfc143f0865943ba72646a1c : SUCCESS in 1h 03m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b6c7b84b015d4c41a9d1526b2a8fce7b : SUCCESS in 1h 51m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13256583a3fd424abf0e4df867ab9e9c : SUCCESS in 1h 33m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ee69f765e2664451b90c441d3b97199b : FAILURE in 35m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07997650abf94e0f8e03a48f7c2e08bd : FAILURE in 21m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b068732cd74d98adffe5c3c6af80e9 : SUCCESS in 7m 18s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/ceea4231a316465d93d84e03a9865f86 : TIMED_OUT in 1h 33m 16s","accounts_in_message":[],"_revision_number":1},{"id":"02565fc0ad677d8e0043d9a837fc58290b47ef53","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 12:38:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77fbd89df5273c55c4d488e945f64a0f8b04c942","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-16 14:10:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c42d0c1d9e696034734b1652a49f43f6ffedfdcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 14:37:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/46d73f5a97064d9f9102b24100671d8f : SUCCESS in 9m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0964ef1439364988929349bd3454e8b2 : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8a5aa47123a546418d5fad3dab2b240b : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f395a6be2e8b4446ad4dd890b87e5102 : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e71d9f0ecdce4de29497beb589f66dc6 : SUCCESS in 1h 06m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9e76f64506ad45a0ac88441007da1f6a : SUCCESS in 1h 47m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4519583ec8243ffb4e298740ab0e800 : SUCCESS in 1h 36m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eac0a57d2d7d40a1b4d61c6b42d99089 : FAILURE in 39m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afaabf799f864942a735e66ea6512e59 : SUCCESS in 57m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11fb6901ca24fc6bac58d188bad5ab7 : SUCCESS in 5m 54s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/f9e52068c34340d994c4475ea48acd40 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8b449878bc28e2ec7c24ea730125f1cb3e455a3d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-16 15:17:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1494cbcd65858eab0ad9105e3aeee46e9c001c33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 15:29:06.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"073724571688ce0f9b27120c49728474ea238ea2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 16:22:56.000000000","message":"Topic set to osa-facts","accounts_in_message":[],"_revision_number":1},{"id":"102c168b8c0e5c35035c936d7a6cc99daa7ea34a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 16:22:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea375064980f2623e00bb8dce9d87db1afe408fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 19:14:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9ccd88d567144e3ea27e5ccd96a0311e : SUCCESS in 10m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/df9e054bdf4c4d618e05f91fd12e2322 : SUCCESS in 1h 10m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d0abc4ed242c4f1c96cb4c869a4e2f4a : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bd474ba523e4d148166922234b95583 : SUCCESS in 1h 25m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7377c9ff6ec54267a9f9e5852526ab90 : SUCCESS in 1h 05m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b9b1dba123b040b7a754362bd90c6ab3 : SUCCESS in 1h 49m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67a7dafa71bc44c99fabf954ea32e224 : SUCCESS in 1h 39m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8d94a9d935024872ad810ea1832365af : FAILURE in 38m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6290c418c2be4b3086184ab8bbd051a9 : SUCCESS in 59m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/450bb0685de147a7a37958bf70a9aa78 : SUCCESS in 5m 37s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/dc12197e3a4b4a96b10accefa9063b1c : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":1},{"id":"79f4da14c159036e25b285bc9dba9ef9c27c3262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 21:15:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab86550ed53e835cf7ce829bcd625f741ec07cb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 22:43:36.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"578ebb7bdffb6ac25b89dedf0343e0c79d29485e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-17 09:52:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"340e07d0a9784b5e5c3080a6146532cc9c84991b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 12:11:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2a0335efe2774b9794c789daa037abce : SUCCESS in 12m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/894993f765bc464ea0f9b91e7c4d95e4 : SUCCESS in 1h 09m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/60710749dd5c4fa489d65630e97001b0 : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/659cbae1175d4ee484738c98c31782c6 : SUCCESS in 1h 30m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18238de5593f4e4bb761c348da9d63d0 : SUCCESS in 1h 02m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3ee810943bc047d098800733358351b9 : SUCCESS in 1h 57m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc67c1a3a58d4f61a9cfc7f215a487a7 : SUCCESS in 1h 30m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/47ab463e979946a88fb1ab7283b6e2ed : FAILURE in 35m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae0f8917f52845bbbb7ba60310a1dd8f : SUCCESS in 57m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c7fd86b2284a499e9a05e1a81743fa : SUCCESS in 7m 56s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/2e2924e6ddeb4d488f3ed4ae59f7233d : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"738dfa9a79abcda39eb66489ec8b1b9e0d038c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 12:13:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a12cf10fbee80e4a321d0ff757f3045c754a5d71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 13:48:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/17b752d91bac4029963a5a6d2ba7d9ea : SUCCESS in 14m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8950e9fd875b416395ceae2a61069e1a : SUCCESS in 1h 07m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/be94449f09154116be38181eaaae55e6 : SUCCESS in 1h 17m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db5a44556ad2428995ffa57fd6efdccb : SUCCESS in 1h 26m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d1357dc429049db93372f44d73ea5f5 : SUCCESS in 1h 18m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6eeae0a9344c4f6e9ce223cc279d1462 : FAILURE in 36m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38b5b91377524adf95af84c280bdb884 : SUCCESS in 55m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/550dd66738d54ec6bf9c531e3e98d415 : SUCCESS in 7m 11s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/a2ca169243e44e94beb84212b7b41040 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ae3b02d99f6a6114d57cc25b4546fa5e15acde72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 13:48:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2e0658b85a3d7edee08bdebfa1aa4291b83f270f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 13:49:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b1a711f1a44ffdbeb2091d1db2e54a : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f249853c0e076a85785851c5f5448cbdc9dc90e5","revisions":{"f249853c0e076a85785851c5f5448cbdc9dc90e5":{"kind":"REWORK","_number":1,"created":"2021-03-16 08:03:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/22/780722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/22/780722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/780722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/780722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/780722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/22/780722/1"}}},"commit":{"parents":[{"commit":"39fe4af6e228fa960eb5281852570afca745fbf9","subject":"Remove remaining suse reference","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/39fe4af6e228fa960eb5281852570afca745fbf9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-16 08:02:53.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-16 08:02:53.000000000","tz":0},"subject":"Convert remaining ansible_ fact variable to ansible_facts[]","message":"Convert remaining ansible_ fact variable to ansible_facts[]\n\nChange-Id: I7eff6aa9bc0e3d635946343b2602ec9431794de3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f249853c0e076a85785851c5f5448cbdc9dc90e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f249853c0e076a85785851c5f5448cbdc9dc90e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
