)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~790006","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ie87b33a672168e9ab54c8d7869a09010a25458ad","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/pki","hashtags":[],"change_id":"Ie87b33a672168e9ab54c8d7869a09010a25458ad","subject":"Add variables for rabbitmq ssl configuration","status":"MERGED","created":"2021-05-06 07:47:04.000000000","updated":"2021-05-17 13:24:21.000000000","submitted":"2021-05-17 13:22:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790006-osa/pki","meta_rev_id":"c3636e214fd90395f0def6fbacf6530ea91b7022","_number":790006,"virtual_id_number":790006,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 13:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-05-14 12:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-05-17 11:35:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-05-17 11:35:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 09:02:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 12:59:14.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-05-17 11:35:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"3c75bb70e8c8442d7511e9cb4a41d36c4c50be1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-06 07:47:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d644df1996cc0511f55cd3cf6a2b9afd58846e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-06 07:54:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e49da9c9f309daffa53e8c5b980102b495fb77f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 09:02:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ae18cb40cf794304bc2259494570f773 : FAILURE in 12m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6002032f2afa4a70a135488118a05768 : FAILURE in 30m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fb9e4cf82acc4cb79aed21b4c1566e1f : FAILURE in 35m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ada9ca65b844fd09e47ffba769434fb : FAILURE in 36m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc8215543a7a45d180a6f6e00e112f9b : FAILURE in 32m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b72995c5d5af4a6884d7b9d40aa7e397 : FAILURE in 32m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/459a599e0bf64daeb7bcfa559b01d8b6 : FAILURE in 1h 04m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d84e6fc66d114ffc8b5adf79dc8e8f3f : FAILURE in 1h 03m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bb2faa7808d4b8098b863e9bfea10bc : FAILURE in 24m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16381ae93f87410d87d6fc3bd119839e : FAILURE in 30m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f796d4abff9c45a6a013e8e277057ea3 : SUCCESS in 5m 40s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/cda6bb7184cd4d80ab73a2fc5303a01c : FAILURE in 11m 17s","accounts_in_message":[],"_revision_number":2},{"id":"072c1a5241eee58603f4b556600ce976379378e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-06 10:13:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"28e9e65594986f34a973f9a38e6d45afadb071b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-06 10:31:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e170107b0cb47e5cc655cac4ff2bad8d4b04d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 12:17:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/14c3de6b1ff04584bb4c05556ba079e3 : SUCCESS in 13m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f538736907864f909ef88e7a3e08f802 : FAILURE in 40m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/018ec5b3071a455d97a37d17347fa082 : FAILURE in 44m 16s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe6c1ec949b940d98ebe03f7a44b4c73 : FAILURE in 48m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d28b367845f7469cb6457ab82848d034 : FAILURE in 41m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/371285852c59487a8119ac983cf1179d : FAILURE in 41m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1ed1fa89fc2e4232bef1ede09f25942a : FAILURE in 1h 31m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcdbabb36558461a8e51bc91b0b5cc35 : FAILURE in 1h 00m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/99b660e44c90495f941b42433a9cfe1d : FAILURE in 38m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b3b1b5bc5a54905bb4dfcc5dba13ac5 : FAILURE in 35m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe7f04f05e343508e06ae1a361b6714 : SUCCESS in 6m 20s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b35151796c6d4ab2b5223502c2dc1957 : FAILURE in 35m 58s","accounts_in_message":[],"_revision_number":4},{"id":"332885592ea6fa3fe8e027f0ec3e67900795ecaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-06 14:15:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8919e125406731887cf530fb7578c77ef4249b47","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-06 14:50:56.000000000","message":"Topic set to osa/pki","accounts_in_message":[],"_revision_number":5},{"id":"a26fc98f235b94183d33858342fa66199ed7cf40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-06 15:13:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"86da7379cab99fab8497dd517f04950c35079729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 17:12:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a1ea9a56204a405dadbf8517f11b82a0 : SUCCESS in 12m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b2c65f50fe3d474792cf9e3f0092d138 : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bb47b74d89ae460e96f8f60f2143b9f8 : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d33809fb979541029aac1d5fd325a2b1 : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b180269288d46a8881a8f34a838329d : SUCCESS in 1h 23m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23fb731cd6804002a717e00f819fcc43 : SUCCESS in 1h 12m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/360feaa6e1d847a48efa95ea877eca1c : SUCCESS in 1h 55m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb209b73e586470b8aa6cc1ca62083a8 : SUCCESS in 1h 44m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/db13189c590f435aa8e2f2e0d9ab5986 : SUCCESS in 1h 04m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c21061e004bb45af8844b4651b54fcf0 : SUCCESS in 57m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01d79d7a82a4262bbf843974694f143 : SUCCESS in 7m 18s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/724f190d30af4c0db111f12fa8bda742 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":6},{"id":"9dca19a25526a45d9f00c6ff7481563b05f223c9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-13 09:09:24.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"a7cde4ce16faf7b320e3496a07716c1a55c143b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-13 14:37:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b576d431d283dac91859899060cb4b98491531a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 16:49:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4d1b0fbd940e450fb2008ce973d67a61 : SUCCESS in 13m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e08646c267864c37b98f6e82917e027f : FAILURE in 57m 10s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1d69ae55355f4bb49e6f30403480fc5b : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd8e6a36421c43238669ce75d1ab40e2 : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68a658345ec94e03aa21c156714b47e1 : FAILURE in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c226e6e30ec84e8bade9876312ee0056 : FAILURE in 40m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/22456590f63145fc96d175c2022b8661 : SUCCESS in 2h 10m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdec6cfd439447a1986b48c03b171c66 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3abe9ca814b7413cb37af67c1a76f5a2 : SUCCESS in 57m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/356f79b7c76d49898ba348cce171d827 : SUCCESS in 1h 00m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e975e145844551895af7c8b4dbebe4 : SUCCESS in 6m 34s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/a4593b09c5294a2f8f6427dee33a1d13 : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":7},{"id":"d13a114d2084b252b8d35689070ac1ed8d3327ae","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-14 09:47:14.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d0a5ff1aed48a24208a0a1d31b45ea596f49c007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 11:35:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cdc76edd4b8b472eb7f6a997a34e5389 : SUCCESS in 12m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2ba35f5b786b447d8b69f9f6fbcd5cca : SUCCESS in 1h 05m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c47badae349d4f8586ea9c99366c214d : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/baf5699d5aea47c2b57feb1783c336c8 : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc123f1d28c8412cb7f79482ecbd361a : SUCCESS in 1h 22m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5f7f323c1774f65abc55bc592592be7 : SUCCESS in 1h 13m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/398cfb9d9b49427f958ff2f229f96023 : SUCCESS in 1h 38m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8328d891a224eaa9e3d15635012290f : SUCCESS in 1h 45m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e4a69f37a374e0f981b839483816e2a : SUCCESS in 51m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaac3a188c0647d18061509a174c7275 : SUCCESS in 58m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e72b6642f14c5eb03faea2f810e9c6 : SUCCESS in 9m 04s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/5c7e7223f11a4261a57cfa3d4ba5fcd9 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":7},{"id":"f3690b6968419c5115aaccc3d490c9cd8fd463d1","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-05-14 12:59:14.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"230b40ca1a25709e9fb476cff2cfb12914d98a20","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-17 11:35:51.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"72bac60ec365b77fde958764d70b1e85147d0b31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:39:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d422774ec7be8b175de7a564b0b7779cce7d7280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:22:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ef5a18896ebf48f7b2943cf31836862d : SUCCESS in 11m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6e1fb2cba92948589d94c7596f9fc1df : SUCCESS in 1h 40m 05s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ddd0190740f94acbbcdc5fb334ac7d60 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/602b70106f5a495c8cb9d492c979a71e : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/483190a4c75e40c5932853265ac8ce89 : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3aa8dfab4884011abd42de8e5ff6b6b : SUCCESS in 1h 01m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0113e15a22c49dda5937aba1937e77e : SUCCESS in 48m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cac2ae3bea84a8dbcf727ae76e87b56 : SUCCESS in 1h 01m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f174e4676c4a83891850c71364b01d : SUCCESS in 7m 36s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/4c8864a6110f4835a65965a826db1569 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":7},{"id":"8e7464472a042203e296752ee4e04e29e4665d03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:22:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"c3636e214fd90395f0def6fbacf6530ea91b7022","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:24:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ff0844959d47a6956090901972721f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ffdcbe7ff834c539ab57e161d82125005e92e930","revisions":{"64efbbc0773cb286efe013831d6fe1edbe4b7959":{"kind":"REWORK","_number":1,"created":"2021-05-06 07:47:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/790006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/790006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/1"}}},"commit":{"parents":[{"commit":"b1f0ee34874ab8d269cf2c46893bb97e990faf69","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1f0ee34874ab8d269cf2c46893bb97e990faf69"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: Ie87b33a672168e9ab54c8d7869a09010a25458ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/64efbbc0773cb286efe013831d6fe1edbe4b7959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/64efbbc0773cb286efe013831d6fe1edbe4b7959"}]},"branch":"refs/heads/master"},"f621a5e14a5ae3a702e44760efc176da5f99fac0":{"kind":"REWORK","_number":2,"created":"2021-05-06 07:54:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/790006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/790006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/2"}}},"commit":{"parents":[{"commit":"b1f0ee34874ab8d269cf2c46893bb97e990faf69","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1f0ee34874ab8d269cf2c46893bb97e990faf69"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:53:59.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: Ie87b33a672168e9ab54c8d7869a09010a25458ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f621a5e14a5ae3a702e44760efc176da5f99fac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f621a5e14a5ae3a702e44760efc176da5f99fac0"}]},"branch":"refs/heads/master"},"ebc8ac0949abdf3e5f1a4587054acb9dd4d78258":{"kind":"REWORK","_number":3,"created":"2021-05-06 10:13:39.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/790006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/790006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/3"}}},"commit":{"parents":[{"commit":"b1f0ee34874ab8d269cf2c46893bb97e990faf69","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1f0ee34874ab8d269cf2c46893bb97e990faf69"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 10:13:21.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: Ie87b33a672168e9ab54c8d7869a09010a25458ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ebc8ac0949abdf3e5f1a4587054acb9dd4d78258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ebc8ac0949abdf3e5f1a4587054acb9dd4d78258"}]},"branch":"refs/heads/master"},"a74774f43461e20064fb179057c8cad471e70a6b":{"kind":"REWORK","_number":4,"created":"2021-05-06 10:31:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/790006/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/790006/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/4"}}},"commit":{"parents":[{"commit":"b1f0ee34874ab8d269cf2c46893bb97e990faf69","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1f0ee34874ab8d269cf2c46893bb97e990faf69"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 10:31:14.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: Ie87b33a672168e9ab54c8d7869a09010a25458ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a74774f43461e20064fb179057c8cad471e70a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a74774f43461e20064fb179057c8cad471e70a6b"}]},"branch":"refs/heads/master"},"d1a3392d0ae3ca446edd9503cd17ece09a99a68d":{"kind":"REWORK","_number":5,"created":"2021-05-06 14:15:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/790006/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/790006/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/5"}}},"commit":{"parents":[{"commit":"b1f0ee34874ab8d269cf2c46893bb97e990faf69","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1f0ee34874ab8d269cf2c46893bb97e990faf69"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 14:15:01.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: Ie87b33a672168e9ab54c8d7869a09010a25458ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d1a3392d0ae3ca446edd9503cd17ece09a99a68d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d1a3392d0ae3ca446edd9503cd17ece09a99a68d"}]},"branch":"refs/heads/master"},"8671ca03c4e6e8e0661b83a403391abd8dd2b27a":{"kind":"REWORK","_number":6,"created":"2021-05-06 15:13:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/790006/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/790006/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/6"}}},"commit":{"parents":[{"commit":"b1f0ee34874ab8d269cf2c46893bb97e990faf69","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1f0ee34874ab8d269cf2c46893bb97e990faf69"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 15:13:33.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: Ie87b33a672168e9ab54c8d7869a09010a25458ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8671ca03c4e6e8e0661b83a403391abd8dd2b27a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8671ca03c4e6e8e0661b83a403391abd8dd2b27a"}]},"branch":"refs/heads/master"},"ffdcbe7ff834c539ab57e161d82125005e92e930":{"kind":"REWORK","_number":7,"created":"2021-05-13 14:37:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/06/790006/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/06/790006/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/06/790006/7"}}},"commit":{"parents":[{"commit":"b1f0ee34874ab8d269cf2c46893bb97e990faf69","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b1f0ee34874ab8d269cf2c46893bb97e990faf69"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-06 07:34:24.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-13 14:37:02.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: Ie87b33a672168e9ab54c8d7869a09010a25458ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ffdcbe7ff834c539ab57e161d82125005e92e930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ffdcbe7ff834c539ab57e161d82125005e92e930"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
