)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~799146","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fwallaby~Ie408aee43116feeaf99ad05cebe764bb855a31c2","project":"openstack/openstack-ansible-os_glance","branch":"stable/wallaby","topic":"osa/pki","hashtags":[],"change_id":"Ie408aee43116feeaf99ad05cebe764bb855a31c2","subject":"Add pki role to local a-r-r","status":"MERGED","created":"2021-07-02 09:25:35.000000000","updated":"2021-07-03 00:23:38.000000000","submitted":"2021-07-03 00:22:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799146-osa/pki","cherry_pick_of_change":799184,"cherry_pick_of_patch_set":1,"meta_rev_id":"9686daae72ebbe03b8b4b1f482b224a8354b6644","_number":799146,"virtual_id_number":799146,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-03 00:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2021-07-02 14:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":1,"date":"2021-07-02 14:11:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-02 11:25:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-02 14:09:46.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"f5e2a021c1668da90ef11348ffc044c8ccade6a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-02 09:25:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a1934991d96a96ac647de2081e877ee734d399da","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-02 09:34:37.000000000","message":"Topic set to osa/pki","accounts_in_message":[],"_revision_number":1},{"id":"e19f78edd541c69d7c490a2dae6114c0b5ab1971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 11:25:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68c0d7f2e00f49a1ac7021b5c95ac252 : SUCCESS in 15m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1c4656ead72e4f31be26b54691727b26 : SUCCESS in 1h 18m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f36b9cc3ace145e996788a6f357d1a42 : SUCCESS in 1h 16m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/de0e162b01e2431e97b1b4700184e388 : SUCCESS in 1h 28m 28s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0e5626f287443929a106488981b7927 : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcbc7242da6f4e58b39ee323a29a0b42 : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/248c98d623fa440f92c424713bb80e83 : SUCCESS in 1h 11m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5013a5ffd77a4cef8f37b07395b966a8 : SUCCESS in 1h 53m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02942074bd1c451b8292db630b295e71 : SUCCESS in 1h 53m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1a874b5fee84ac593b74b6aaa0cd7d2 : FAILURE in 39m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04d2fc22f2fe4f9e9c45419033e8d509 : SUCCESS in 57m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b85e9735beb4cbc8c883378e3dc1e89 : SUCCESS in 6m 33s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/cc2e4caf71d34767b63ed04a610efc10 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3c5564bcfa5a99f916f69204d28f73eda9d71bda","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-07-02 14:09:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b977c9ca94dc7df065bd90751c56dfab110e3ed6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-02 14:11:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9d77c4b7b805b8ce2b9985e4c1e2891decb5828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 14:12:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8886f2bca352f9578147b3e0680bb0cdf2bb14c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 14:33:45.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c253a588de70debc8aa1a20acbd7d6c088fbdfdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:07:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5174aec5addb432681c33ed3c9882eb4 : SUCCESS in 16m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1d8207cedebe4db48d3ff2357e92b164 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/53dbcee9f9c34c20aab73f5c45e6b9da : SUCCESS in 1h 23m 53s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/17c713622bd549c3b46b2699c7d4a976 : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39650222a9984825b5891f6e457b7398 : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74c91f729fce4f9691b71db4577b4776 : SUCCESS in 1h 06m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2358a04de320479786b6e619f21d37bc : FAILURE in 41m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8418eb4669374f64b7d81adacc8ef154 : SUCCESS in 8m 30s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9806d710b4a94d82a441cbc508f5e915 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7a294b8a374d357bcc1dc79da322e05cc335a16a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-02 20:37:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"359984404683d7e2226f27eebf1a30358b70e6ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 22:32:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f01c644e57984341a99f65bd8d84a1e9 : SUCCESS in 13m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/097b8063d9504ceab780cf4afb945426 : SUCCESS in 1h 09m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/36081f02966d47b185795f1f1ef44958 : SUCCESS in 1h 07m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d5efdfd91ca842a9b7ee2a5559e45c6e : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/05e2f1685c904c8795d94f194e5a59af : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af8c3f7d3f3a4329a0a304d2c6974040 : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c217b94b4644d7aaee7e0f6efa79ced : SUCCESS in 1h 10m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7c24383892db4af29ec549c15f339631 : SUCCESS in 1h 49m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e551c795f28f4c6bb917e5d4dd21faf8 : SUCCESS in 1h 38m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/91bc3cd13fd148ee9a66811b6e03e385 : FAILURE in 40m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a74014553e44a2e9736f4265b54afea : SUCCESS in 57m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd11f75ab9143d8811a1b2b4e0641a8 : SUCCESS in 6m 10s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/05c9e853067641e9b96b2dde9dee776a : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cca71f36af4aa267d2071a24e0da36417b0f2086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 22:32:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"192659b7aa6e97057ca92e07610f3094fcf1721e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 00:22:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eb974d2231849ac9c2a679a60da0ce2 : SUCCESS in 12m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/37eed1ca5d3b4435bc5ffa38f1679451 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1527c22947a643379781b5eb32c5570a : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a3296b00b5a4b7aa0732641eccf2e34 : SUCCESS in 1h 21m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12dcb575fb88464c9807aeac729d6596 : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f046d25603424b53b1a6f9640045fd75 : SUCCESS in 1h 47m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bcd0d1f826e47389773635d31494736 : SUCCESS in 52m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6707ee0a7da440ebb707b9d01d2b676 : SUCCESS in 6m 05s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/93c598bc647148e6baa343c9bc009157 : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0b08a8ce816a72880cb5ac293f1939b8346da077","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 00:22:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9686daae72ebbe03b8b4b1f482b224a8354b6644","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 00:23:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff6e83d4d4d34af9a2c114265f7d7ec0 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abf44fcaa65ba774d25f638a0941562278eb20ad","revisions":{"abf44fcaa65ba774d25f638a0941562278eb20ad":{"kind":"REWORK","_number":1,"created":"2021-07-02 09:25:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/799146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/46/799146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/799146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/799146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/799146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/799146/1"}}},"commit":{"parents":[{"commit":"09e142ce94d32f001bdcb6953cb6af80ae3985ae","subject":"Update .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/09e142ce94d32f001bdcb6953cb6af80ae3985ae"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-07-02 09:19:56.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-07-02 09:25:35.000000000","tz":0},"subject":"Add pki role to local a-r-r","message":"Add pki role to local a-r-r\n\nChange-Id: Ie408aee43116feeaf99ad05cebe764bb855a31c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/abf44fcaa65ba774d25f638a0941562278eb20ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/abf44fcaa65ba774d25f638a0941562278eb20ad"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
