)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~809635","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I71ebc2fc4e386f3a1599fe73d49fae185ec9d2ff","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/galera_pki","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-11-11 11:01:36.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"I71ebc2fc4e386f3a1599fe73d49fae185ec9d2ff","subject":"Refactor galera_use_ssl behaviour","status":"MERGED","created":"2021-09-17 11:39:13.000000000","updated":"2021-11-11 14:45:50.000000000","submitted":"2021-11-11 14:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809635-osa/galera_pki","meta_rev_id":"36208e3d887990599da9d5324281b68b25daeb42","_number":809635,"virtual_id_number":809635,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 14:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-10-06 09:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-06 11:15:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-06 11:15:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 13:50:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-06 09:19:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-10-06 11:15:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"39289d7237b4291d2c37a9c2a4d61807aad7dd07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-17 11:39:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5488764a2896af0e4449755118300e39875cb6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 13:50:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/119f663ad55e4f5d80f2104e929afbd4 : SUCCESS in 14m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/503daaaa731341d6b8448200c4905db1 : SUCCESS in 1h 15m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3ab61dc46e234a6989edeb774ce7de37 : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7cc5a75a079a49b6a7bf2fd9d18c1c12 : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9babd5795d644210a2f639cdef50e66b : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bfa5d431f8f48bc8d1274146f4e5c10 : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d0c8fa7eb0644fab8c3c60055450d66 : SUCCESS in 1h 12m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6118cf326474b499d30b90fe375e4cc : SUCCESS in 2h 01m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d29e608ec3514fda8984bfdb36aca080 : SUCCESS in 1h 37m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6840941e46e4983bf6ddde595b933a3 : FAILURE in 41m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b26c8b14b7f497d985632719766db3c : SUCCESS in 1h 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3609a150585c447fb7b9d080f6432943 : SUCCESS in 5m 57s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/9e431d4fa97c4a36bad5264a4f2fd182 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a6a470df61216abd7efe3d4eceb2a1f619da6537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-17 13:58:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"89beeba1000e6bcb0e21722450c5d6a38995a168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:14:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d0432df351e4ede96ad8cfe543fcdff : SUCCESS in 15m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0f9043c5c6474506833dd819e8e3d0ae : SUCCESS in 1h 53m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/21628c17cd4b4f4583728bbdabece2c4 : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c927e8ab5aa64198964675f52618a8b2 : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/99ba8ba3d37d4e2baad7e354a4a29c0c : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/506107dcd1b44e178b167e76d143aeb1 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0ec83c38f4f4b349e1a331aa531b50a : SUCCESS in 1h 12m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2037e2863cfe4f0f986ef5358eb969df : SUCCESS in 2h 06m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caf87cca8dc142e5bbb93b3e31f3f74c : SUCCESS in 1h 36m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/257d12ee6f4040eeb023b79078f4bc6a : FAILURE in 52m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecef870900954e69af273cd0a9a44298 : SUCCESS in 1h 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/953437bfb40c4687b2f7b41f19b1e52e : SUCCESS in 7m 34s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/b6d33b2be1554f3c9ff8615949c9e9d4 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a4175ac896d70fdccd9af50d81eaae18c447562d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-06 09:19:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"db8ca74898f21d4125cd15f8518a1dc953204a28","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-10-06 11:15:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"24fc28730121fde9456738afbf3c7ba54cce96d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 11:17:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5645056be5ee004199492901dde98c5e7c392528","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 12:45:29.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71e1c07c8e234bfc9e623357cb257ab3 : SUCCESS in 19m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6f6ba7162371487981f148aafe5a7b59 : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7e231379b61f4d0a8f71b00cd3f0defb : SUCCESS in 1h 11m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dd81769082a840b886e36a98c7fadf14 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/487d9c4aaab14c02b09fb4b847f65837 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f12f682213d4433a141c84aaa0d9e58 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f13b7b39e33543c3bc170ca340a81f45 : SUCCESS in 1h 11m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e18392a7904f4ca5897f6e0104b05a54 : SUCCESS in 57m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fbb7b34520439c866887b7b260c4ea : SUCCESS in 9m 27s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/5d623253192746f4b5a05bdc85c2e4d7 : FAILURE in 22m 36s","accounts_in_message":[],"_revision_number":2},{"id":"747f60400f8489083d9feb0def2ec3f6ef4d1316","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-06 12:49:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ef1d8e4c75073fc4121a1013f9fbdcc63b9f6f29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 15:06:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/add9679cf0c246cb9648bd7899ee0a2e : SUCCESS in 12m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8dfedce1e5644b58a03a4636634a20f4 : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9c9612ae02b849bc9f68b9a0e42d2f7b : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5d79b09abe354d75bea4d0ce87165cfd : SUCCESS in 1h 30m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/61fdba44a9e847ce8059afb2115f22ae : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acfae7f088c541c090bb700ce1e48431 : SUCCESS in 1h 24m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5d0c128033c4cf28999e880137768c5 : SUCCESS in 1h 10m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/be027e88ab0647969b965b961d259761 : SUCCESS in 2h 09m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f54073b1e6734771be5c59c75931262a : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdcb64830a9945c68da5811b859331db : FAILURE in 43m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6ce3182c53f412bb6623b8029777c1a : SUCCESS in 53m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995315a7c3814c0ba8969dce2ab88d5a : SUCCESS in 6m 18s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/3e44b6173aab4e30a5a388948ca038de : FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":2},{"id":"936679d33edbf86192d8b53748de41bb53d74426","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-10-26 14:09:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8f5aa6417dd5a44e292e941f4441f521dd3c05d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 16:12:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a86bba0eab724693bac83d20732034d3 : SUCCESS in 18m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8f4b5c81451145c2a1d1d4fa26c8036a : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5e73fe47cfa7438381d77da734220c1a : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/97e6955d7bd34f8a83c728f9366f4322 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/db3310db100c4f7992264ce561517acc : SUCCESS in 1h 26m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ac2270bb85d4fe1af3ee1c6d2a7c853 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a634ce675954459ae31e04e5d8c36db : SUCCESS in 1h 10m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e1be20104b543a9921da109df8fba1a : SUCCESS in 1h 55m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfac44cce03d4504ab7128061a17baed : SUCCESS in 1h 35m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fca96102b83945b4bd9724247dc14223 : FAILURE in 43m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b14f0275bb75476cb883968dc4da13d9 : SUCCESS in 54m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22570110ae440288715980ad77058f6 : SUCCESS in 6m 04s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/7fddc09c259d4d73ae404d7182b83900 : FAILURE in 21m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1fd55ff773436e6954840e925aae3bb860b5a1a9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-05 14:12:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38f0c97b38871e48f763ac9110b79c7a62819816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:31:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4bb333bf53845699a48f3c73dbe8c6e : SUCCESS in 14m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4ea009af82f941b9acc540a796473775 : SUCCESS in 1h 19m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/30a92a8d66f44c0c804ffe7209119fcd : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5beb2c67354e45a5be0602c7de897280 : SUCCESS in 1h 20m 10s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b40a980d7a8a4cc9aceb5a5de4f69f0c : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9527e9b7ebc4481aa37a339b6a62d45 : SUCCESS in 1h 22m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9de11438514f4b55baf3404b91ee6a88 : SUCCESS in 1h 09m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/319d6a07a3624eab8fae3ffc0d28654b : SUCCESS in 2h 11m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26dde2d959954f2687cb9c21dc219774 : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7a8267454a54ed2aefc01e35b0f2efb : FAILURE in 42m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5baec442fd944651aaa806b398f4851f : SUCCESS in 58m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfe4db9c7ba413b8e15c6e877c5b8d1 : SUCCESS in 6m 12s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/0cc0cbac88db4085a5bb10bf0b830aa3 : FAILURE in 21m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6fa2b4f9811039528e66155c12fb2fdafe8de6ef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-11 11:01:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c73439a76dea826872129200c3b530d3e564946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 13:05:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ada7f2787d439bbff40926a7862558 : SUCCESS in 21m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3cb4f5204a5a44e8b47e267546f4426a : SUCCESS in 1h 28m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/96511c48248f47219108ac8797defb68 : SUCCESS in 1h 08m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/999401bb82204907a424917d2525849d : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ed41de46e414b7a89b011610a2f557b : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2ea2f68436841a19709857326322b31 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acab4a7370d84ff1baa9247a64cba675 : SUCCESS in 1h 15m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b4d2a4ba29c409b9b4a675cae8f277c : SUCCESS in 1h 58m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f69240f242524b2bbc755f35d9fada11 : SUCCESS in 1h 43m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/720ae8b1a5b543c695552e32714666e0 : FAILURE in 45m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d8e970a82a34a6aae4fb346b1295abd : SUCCESS in 53m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd2aaffd562457eb1db57312051deea : SUCCESS in 7m 25s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/0ac6482697c144bd8b2d0811a79d882f : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":2},{"id":"15bdfa5af3bb9bb57b1823f2030a9c8c3493a3ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 13:08:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45b9d0cc308cf490252e085b067e747334348ac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 14:43:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/103fa2c29b254e22ba8fbce629d0b052 : SUCCESS in 14m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1d78765427f64fa18200718cdf13b755 : SUCCESS in 1h 04m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6fcf9a8129c14d18ba0be6a9ec5b153a : SUCCESS in 1h 00m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/29c16c83ae18409ea810e02091cb1e68 : SUCCESS in 1h 24m 00s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a92dac40eaa24c3281038566b6b28c26 : SUCCESS in 1h 25m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/906891cf4f7e4cb58c9b696e2f87b354 : SUCCESS in 1h 34m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22904277a21847c18a60356ee3d67975 : SUCCESS in 1h 11m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9de09dcade344cf29cd363ef9a3b278d : SUCCESS in 1h 01m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd41ac26312d4ceaa87ed96c15627859 : SUCCESS in 5m 40s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/3a6ecd3c79c5483b9c637a4e1c9fad21 : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ca621748d8b1f79d7cb8b07204dd09569a9f4eea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 14:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"36208e3d887990599da9d5324281b68b25daeb42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 14:45:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff10b19572e54e0488f36a8fd739b98e : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b719c7ffb3554576aadb0da9b2f655659c86b15","revisions":{"ec6638d000fdbdc98ece1ce7d86c8af7775190e0":{"kind":"REWORK","_number":1,"created":"2021-09-17 11:39:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/809635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/35/809635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/1"}}},"commit":{"parents":[{"commit":"afd07792015fa6452eb24b3d9a1d4becc8240922","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/afd07792015fa6452eb24b3d9a1d4becc8240922"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-17 11:39:02.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-17 11:39:02.000000000","tz":180},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: I71ebc2fc4e386f3a1599fe73d49fae185ec9d2ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ec6638d000fdbdc98ece1ce7d86c8af7775190e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ec6638d000fdbdc98ece1ce7d86c8af7775190e0"}]},"branch":"refs/heads/master"},"7b719c7ffb3554576aadb0da9b2f655659c86b15":{"kind":"REWORK","_number":2,"created":"2021-09-17 13:58:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/809635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/35/809635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/35/809635/2"}}},"commit":{"parents":[{"commit":"afd07792015fa6452eb24b3d9a1d4becc8240922","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/afd07792015fa6452eb24b3d9a1d4becc8240922"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-17 11:39:02.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-17 13:58:42.000000000","tz":0},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: I71ebc2fc4e386f3a1599fe73d49fae185ec9d2ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7b719c7ffb3554576aadb0da9b2f655659c86b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7b719c7ffb3554576aadb0da9b2f655659c86b15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
