)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~821097","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Icf8f2eb3b27033597cb75a6c990b7492706f1d05","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-10 13:45:57.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"Icf8f2eb3b27033597cb75a6c990b7492706f1d05","subject":"Replace NFS test with integrated one","status":"MERGED","created":"2021-12-08 16:44:48.000000000","updated":"2021-12-10 18:51:53.000000000","submitted":"2021-12-10 18:46:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821097","meta_rev_id":"439a2465a67434b85677a5baf9e2d5c4269ab273","_number":821097,"virtual_id_number":821097,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-10 18:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"recommended":{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},"all":[{"value":1,"date":"2021-12-10 10:21:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},{"value":2,"date":"2021-12-10 13:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-10 09:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":0,"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},{"value":1,"date":"2021-12-10 13:02:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":1,"date":"2021-12-10 16:20:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 19:03:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 09:19:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-12-10 10:21:25.000000000","updated_by":{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},"reviewer":{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},"state":"REVIEWER"},{"updated":"2021-12-10 13:02:31.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"}],"messages":[{"id":"b723f85f79c7e19a0feae69b63699e7bd9903c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-08 16:44:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79ba544eb4a8ba164d9d00d09be2f8b0280dd3f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 19:03:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99a8349ade8349439d3beff75af0642c : SUCCESS in 17m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b829a599077d4ab6b060ef231aa785b0 : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/16da18fb8c19417a81e5a8709fe60ef4 : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/890707262bb842e8b2a4e9a7176d9c39 : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f192e4456574ace8d0abcf6761ddc25 : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daece0386b2e45f5bfab5bec9cfeb054 : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28573668d03c48979d7df817ca062610 : SUCCESS in 1h 26m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a9b9e0b331049ce917829346a43e10f : SUCCESS in 2h 04m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47d45212da744ad9bdffcd115ff47ffc : SUCCESS in 1h 52m 21s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dc28ec0f73b4ae585163d0e4bc9e3e8 : FAILURE in 53m 59s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e3d67790d844466814844aff9544309 : SUCCESS in 2h 11m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d8c9109d4c54d43be785f2fbbb9144a : FAILURE in 42m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faaf3113c67a4e8db3f12102240b6b0a : SUCCESS in 54m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e2061aebc546d08aa902bdf0f95646 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0c084f41a534534ce3f2ef97d244dc7fc77c6485","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-09 13:41:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6903d2fded6c0f7d49dd7512e534ef9930a97328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 16:47:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0269be8ff41941618727ff893326fb5b : SUCCESS in 14m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fb91b451e3b74562a14075281620a128 : SUCCESS in 1h 33m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1b9f10ba162548699de6790955e3eae4 : SUCCESS in 1h 16m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2019b5807251421eb4b91aa24c911efa : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d18c154d8fb046a38c82021a57a60160 : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2111144be7bd42d4b09f36b575795a72 : SUCCESS in 1h 35m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e38e323c0cd4bd293eb39c4d8ad008a : SUCCESS in 1h 20m 16s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8209a31b0d9549ed9bb44b4578ab0487 : SUCCESS in 1h 49m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a03a6d6aaf4fc3936e914a119c4e71 : SUCCESS in 1h 56m 49s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e0aa8cad3b44379bdd90645eec0ae17 : SUCCESS in 3h 01m 31s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee57bb93d71b4a37a412f565a25b134c : SUCCESS in 2h 03m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/426798aaae934e06822792902bc48fbb : FAILURE in 36m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33311a7ddb1944b59a75d121c595463a : SUCCESS in 54m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c256e8a46c54475831ffc97b5449ce0 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6c0e80f03a9e9afc3df6e345a451d088010a9e5b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-09 17:48:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"163b31cf6730737f9b7a54f812638bd5fd051ef9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-09 17:49:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"08b0f8cd461476aca9bc22315065dc736cad9165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 20:12:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f2a5013e0f143e6bfff308c38e679e5 : SUCCESS in 16m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2f6c5ad97b3746f58839a235ed90a0d3 : SUCCESS in 1h 39m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/11e46d793658409c811f3004291be89d : SUCCESS in 1h 19m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4690a0d7c41140758d1e6f4d6704dcec : SUCCESS in 1h 29m 44s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c94db44da4934c82bc5493b7b28e074c : SUCCESS in 1h 27m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15794bd0e2834f06991a8e8f16488683 : SUCCESS in 1h 37m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3442142d40849ca8332e5e433ba3724 : SUCCESS in 1h 26m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/884a12259d1046c7bf55bcae288b1d7e : SUCCESS in 2h 02m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bda1cc3fcefd47e385ebf2eda3ecfb4d : SUCCESS in 1h 53m 29s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc82e51163ee43699f049dcfbf0b5810 : SUCCESS in 2h 00m 36s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355c5d7ff9e24d588248eed42423386b : SUCCESS in 2h 17m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/064045570afc49ba81636837557078f6 : FAILURE in 39m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e586764d5580410ba102fd6fe6788cd6 : SUCCESS in 1h 01m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5006b318204bac8c001c3236eff4f3 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"55ac31022b2dfc122bb487fcca3d8be693462649","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-10 08:37:16.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"635224e253f51c32392a707c108a9439f764a578","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-10 08:37:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8f16b888f03f10ed0ee50a20ccc88756c64fa0c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-10 09:19:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85bb9144f2a21ce1bedbbfe3a3d5cb74ec5d66e4","author":{"_account_id":31749,"name":"James Gibson","email":"james.gibson@bbc.co.uk","username":"jamesgibo"},"date":"2021-12-10 10:21:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"04bbde295fd46039ff8bdc871ad979c6b3967918","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2021-12-10 13:02:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8d31005e19d387e676c8b8228f4d16c716504d4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-10 13:45:57.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"192a529f4a9023d99a66175a46dc5775aea3942a","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-10 16:20:43.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"9ac61f21ef85a30aee6805accbd2d15fa3508034","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-10 16:20:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"270c3cd577a51d28a4d69404718b36189a6ad0c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 16:23:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5426da176a7ad23c8d4b36486244bb95cb22c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 18:46:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d89ffe10431d4d6b8e29f804e48a7fbd : SUCCESS in 15m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c6a64e61dbf64797b432d4a3af5d1a5c : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6529003101034d6ea3cf2fa832d76c59 : SUCCESS in 1h 30m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/91a7d629a3a34bcf96b2663c7a214828 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9043ebe885740c0b9e0ab370ea3f053 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9318d2ca85454c76a4feeca0cd306e7e : SUCCESS in 1h 50m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d82708fc7faa4455b2cf17eff93a786e : SUCCESS in 1h 30m 08s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaac9d983105470a90e1ec6c769c191f : SUCCESS in 2h 16m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97669cf9639440acbc8f27017573274c : SUCCESS in 2h 20m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d97d9aae326418dabc7c95d0a631c45 : SUCCESS in 1h 00m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a404a221856c434f9e00c21ced8687b4 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"78487f94fb1c5e2d3bb9c37bb11a404382c7dda1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 18:46:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"439a2465a67434b85677a5baf9e2d5c4269ab273","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 18:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4e0180a0c84af1b309035d92a2141e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebe439abb44e3759f502649919e5567dedb874ec","revisions":{"ebe439abb44e3759f502649919e5567dedb874ec":{"kind":"REWORK","_number":1,"created":"2021-12-08 16:44:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/821097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/97/821097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/821097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/821097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/821097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/821097/1"}}},"commit":{"parents":[{"commit":"868eef21e61714c3905a29cf09704abedf74c60f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/868eef21e61714c3905a29cf09704abedf74c60f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-08 16:44:40.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-08 16:44:40.000000000","tz":120},"subject":"Replace NFS test with integrated one","message":"Replace NFS test with integrated one\n\nThis also adds ceph integrated test usecase.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/821096\nChange-Id: Icf8f2eb3b27033597cb75a6c990b7492706f1d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ebe439abb44e3759f502649919e5567dedb874ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ebe439abb44e3759f502649919e5567dedb874ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
