)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~821214","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ib158e14c6f296795bb1f6eabdcfc23b80cbcd871","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-18 06:15:57.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"Ib158e14c6f296795bb1f6eabdcfc23b80cbcd871","subject":"Stop managing NFS mount destinations","status":"MERGED","created":"2021-12-09 13:56:31.000000000","updated":"2021-12-18 10:45:43.000000000","submitted":"2021-12-18 10:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821214","meta_rev_id":"0fb13579cc79b569a39d37a9dbc6389dffa56c0a","_number":821214,"virtual_id_number":821214,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-18 10:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-12-17 08:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-09 19:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-12-17 08:19:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 18:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-09 19:21:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-12-17 08:19:05.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"de4a1945aa63705103cf1f1f42b95a34a5e00bd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-09 13:56:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e575eb2d237869b59712a8ab528c7cef730d34f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-09 16:14:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2f04f4a76b26dd32a220277cfb4e601038285f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 18:42:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6e6fb59843e4b4e82cb928d2dfa81b3 : SUCCESS in 13m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b542240273bd4386861c80b973f7569d : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e482adac432f4900aa977c81abe599a2 : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b803b0e8618d4f03bc6d9cd4a6fbf94b : SUCCESS in 1h 15m 40s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/08b8897c0f56456b9efe972dd738bc48 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b219a2bcec644d8aad1e00bab27a523a : SUCCESS in 1h 26m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/088f161b81d940cdb07f75c6313a8d53 : SUCCESS in 1h 28m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/020f8b8e8e254883a91c67e87494670d : SUCCESS in 1h 58m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5c18f3dc8494a0d97fe0311359d892f : SUCCESS in 2h 01m 22s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/860c191982904e1d98dbb1fe55304a63 : SUCCESS in 1h 59m 36s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05192cad6fb44c0e9dd8b645b57a94b2 : SUCCESS in 2h 21m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fd88fb4225c481382997c747ce3186e : FAILURE in 37m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0e20d50421a44f6936162a8b56b06e1 : SUCCESS in 54m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b0430b99a749b2a757e892a66ccd5b : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":2},{"id":"2c48b30882b45b98c2dbe8d44a984b1b1fdaa13c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-09 19:21:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dcc65ba969cd0488480896a8bf75e243b9aa4978","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-12-17 08:19:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1658340adabff51017abcc244e07e4c03a50ecac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 08:21:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f5ce3d2af7eaf4e09b62c5291c37e7b1f695313a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 10:24:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af81690d67514ee0a1a2cc4a7309a0a7 : SUCCESS in 13m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/739f61e0bdf04c76b3de2d74165e0d0a : FAILURE in 35m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a0e59b729740459c912b5dac59c77a9a : FAILURE in 23m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5b82db6f918e4170a021cbfae14c0e5e : SUCCESS in 1h 18m 12s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff627526c4124c098e5e8a629640845b : RETRY_LIMIT in 10m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f178be42751c4e639363c91b042493b0 : FAILURE in 33m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6e7a6318cd948a485a532b9f8ebc6d4 : RETRY_LIMIT in 10m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/469b353365074da5ba2df6577a225c24 : FAILURE in 33m 28s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1c77895069546c9982862219b209888 : SUCCESS in 2h 02m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63501263c4174361bba5b865be2e57ec : SUCCESS in 54m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1154349b6bc4229ae76c2e9bc3b1169 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f79e46014602da9440f3c52e0c7bca0d9474e238","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-17 13:29:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dff9b23b1ebe764fe9b59b9d11b032c4db4d9c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 15:48:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f49c1174ffde47509603d332a9c338fd : SUCCESS in 17m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1d328980c0a444bb83274487cab8f3e5 : SUCCESS in 53m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c969830c107341149b6208db5102e768 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/344443bc281b4d258a0bd79b8fdb6c8e : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a98c26c16e5c42d2bce5811824bf6b0d : SUCCESS in 1h 29m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f55c8ba8bf60401ea17ec35007bb45a0 : SUCCESS in 1h 18m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4650c0d0b7be4288940aee4843fefc12 : SUCCESS in 2h 04m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af182dfe54c44df2b8e637f9d50e5e8d : SUCCESS in 2h 06m 30s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bde73362e854113b9a181951a19904f : SUCCESS in 1h 51m 28s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22b6b4f245694f8d9034f00d600ca257 : SUCCESS in 2h 09m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb8cc0efdb67442a89005332b2cc0071 : FAILURE in 22m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10fca5e53c2f4e98b34ddcbb26821ccc : SUCCESS in 54m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7d2f5c6db24d409ea31bc37f1922f9 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":2},{"id":"762efcac31c1b7fe32b5557e5a42b6544564991e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 15:50:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fa3b43ee2e2cb5eeff25b221d23f039b68dfd616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 17:49:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2ff68cdea3e47559281faa63aff9b84 : SUCCESS in 21m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cc332f0f583b463b85613d60537eabd2 : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e46d15ed8cef4623b915f3ee565c424d : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/af32ffdb8f7f47948acaf44091c66810 : SUCCESS in 1h 33m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/821f8b5d8ba4459f84df9451040b8b45 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a754d6e55b264396bd6501b31ff15aa4 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c72eaf202a414d6fb9c8d8faf2e54cca : SUCCESS in 1h 54m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c47e32b53324801a37f2b75901d559c : SUCCESS in 1h 50m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f437c665774242d791195a4d2b9b2c19 : FAILURE in 48m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49a12b139a9402ea7ba716f2dbb41b4 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":2},{"id":"41c90d0a1f30f521440b640165ceca01a60925e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-18 06:15:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f21abe80dd8bada650e9e66a9c298c5fb9c6578b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-18 08:36:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/753737af0bf94ae5be4e307259a39a2b : SUCCESS in 12m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/932bbac67f3e45be99fcab166d7167c7 : SUCCESS in 1h 00m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ce926742a6cc46e5ab104320ce559be4 : SUCCESS in 1h 21m 59s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d972cfdc2af941c6b6e64fe82e13a6c2 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5766494b05574a57a23235502bf1bef2 : SUCCESS in 1h 18m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4570e77b91f745369723243794f8a11f : SUCCESS in 1h 52m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/779a8f04e1844da292ae8322d44560cf : SUCCESS in 2h 11m 25s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1f83f7d6b5044bb8c4702ab9962228e : SUCCESS in 1h 59m 28s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b01ace608d74e13ac4ddcc1ce20434b : SUCCESS in 2h 06m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/64b93d514f564c6d80f599966c382454 : FAILURE in 22m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd5bd05ae1a74df98b5a81193ca50d58 : SUCCESS in 48m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80bd5209ae614fe996f7085674dc7a3a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"223e2c3478374ccfd74f47ea17eb7877969f7c94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-18 08:39:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abace611353352353f459a06c13a772d13de0050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-18 10:43:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d91ff0e2ff34a6abd4b1300e74abd86 : SUCCESS in 13m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/373b2928bc124c6e88bfd8d286f5c707 : SUCCESS in 1h 00m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/92cd7d3d89104ecd98200859c34d814f : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9840b2e8b7943a8904a8f9df2b71722 : SUCCESS in 1h 22m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be8653e0f37b4d199d12189d9cbb70b5 : SUCCESS in 1h 16m 06s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e8355583822415fba70109ee8774c6d : SUCCESS in 1h 50m 50s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ad5d30172d640abbf2bf507a6880e89 : SUCCESS in 2h 00m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a72a0ffbcc84d74a0f25f9f3c44c967 : SUCCESS in 46m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5bc2177a61b4124bf750a289dd8d95c : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"96ee0d166f570825de5d0e845db46aabdb7f698d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-18 10:43:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0fb13579cc79b569a39d37a9dbc6389dffa56c0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-18 10:45:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76831799fb1448aba3d8d444a5283240 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0af66c540f3c73a729dd85a2a9d8919e491389e1","revisions":{"881a587d39cf6afc5b247c52943a7df02e6b9b2c":{"kind":"REWORK","_number":1,"created":"2021-12-09 13:56:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/821214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/14/821214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/1"}}},"commit":{"parents":[{"commit":"ebe439abb44e3759f502649919e5567dedb874ec","subject":"Replace NFS test with integrated one","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ebe439abb44e3759f502649919e5567dedb874ec"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-09 13:56:17.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-09 13:56:17.000000000","tz":120},"subject":"Drop NFS destination mount points","message":"Drop NFS destination mount points\n\nSystemd mount role now takes care of creating mount destinations\nso there is no reason to handle this manually.\n\nIn addition to that, if user/group were not provided explicitly to the\nones set for NFS export, role was failing with permission denied\nduring re-run\n\nChange-Id: Ib158e14c6f296795bb1f6eabdcfc23b80cbcd871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/881a587d39cf6afc5b247c52943a7df02e6b9b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/881a587d39cf6afc5b247c52943a7df02e6b9b2c"}]},"branch":"refs/heads/master"},"0af66c540f3c73a729dd85a2a9d8919e491389e1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-09 16:14:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/821214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/14/821214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/14/821214/2"}}},"commit":{"parents":[{"commit":"ebe439abb44e3759f502649919e5567dedb874ec","subject":"Replace NFS test with integrated one","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ebe439abb44e3759f502649919e5567dedb874ec"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-09 13:56:17.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-09 16:14:42.000000000","tz":0},"subject":"Stop managing NFS mount destinations","message":"Stop managing NFS mount destinations\n\nSystemd mount role now takes care of creating mount destinations\nso there is no reason to handle this manually.\n\nIn addition to that, if user/group were not provided explicitly to the\nones set for NFS export, role was failing with permission denied\nduring re-run\n\nChange-Id: Ib158e14c6f296795bb1f6eabdcfc23b80cbcd871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0af66c540f3c73a729dd85a2a9d8919e491389e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0af66c540f3c73a729dd85a2a9d8919e491389e1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
