)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~822870","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-28 15:04:28.000000000","reason":"Change was submitted"},"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2021-12-28 15:04:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa","subject":"Add boto3 module for s3 backend","status":"MERGED","created":"2021-12-24 03:15:45.000000000","updated":"2021-12-28 15:06:14.000000000","submitted":"2021-12-28 15:04:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822870","meta_rev_id":"46ef64a840cfa720d5b92d033db75d6a56e97776","_number":822870,"virtual_id_number":822870,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-28 15:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-12-28 15:06:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-28 15:06:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2021-12-28 15:06:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-24 05:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-24 13:02:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-12-24 17:30:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-12-25 09:03:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"9ba2b4d944a3e1cad2810a14e15fa5eb0dc5b147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-24 03:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7be545533d48cb5be8bbb9826613d735e3af613c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 05:18:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9265ed4a51384692b3e373d947702152 : SUCCESS in 15m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/09b568207f0f4da0ba9503153356a1e6 : SUCCESS in 52m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/333a7ea899a24200be5807ed774cdc2b : SUCCESS in 1h 01m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/776866a6df0840faa304c37bf813d1a8 : SUCCESS in 1h 04m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43d9461fe95944b09d36a804378ad73c : SUCCESS in 1h 05m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/30bd0a22e3354c52912b8e73da8e7c4d : SUCCESS in 1h 58m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b77211ba942346b4942bd76f819af879 : SUCCESS in 1h 50m 03s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d70671fa64746f889f0cbea68267384 : FAILURE in 41m 33s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c377e60bc0ae4cf983c88e986a7b94ce : FAILURE in 33m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1aac6347e3b84cfc89b4ec37bac1e77f : FAILURE in 21m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/146178e7cad2417f9329dde111508dd9 : SUCCESS in 42m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde5f125434a47babf7bd89d8e42fb9c : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"305784533a2884076e8f34a2f3bec300448f62fe","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-24 05:25:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f426a0080afc3a51d71fa0249a65ffc12e7c7694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 07:25:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e00b228a9cc547cfbd61f77b283d21f9 : SUCCESS in 17m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/29e032dd41ee4e9bbba5fabd7e06179c : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b3ef3fb3def647c09618086d2608902d : SUCCESS in 1h 06m 12s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a30d80c86bd847d293d01f9876bbb18f : SUCCESS in 1h 14m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78a21b93ca0e4c1498de91dbe41a5363 : SUCCESS in 1h 19m 50s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b72ff373bd8b4ca48dc375c10a59db78 : SUCCESS in 1h 54m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e78b7c55f69348129a4656da2ec7177c : SUCCESS in 1h 47m 28s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edd27cff6a74428ea837db7d90b0e152 : FAILURE in 38m 16s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db3508a6cdbc43fc8cea3863ba605d6f : FAILURE in 39m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e17936200b14989944eff1a8858b187 : FAILURE in 27m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eff7540ddcac4edd93e9e09701d4376b : SUCCESS in 51m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4f7979975d452c88274f0ec1c584c2 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"19b501646ff4b96784c1b2dfb3dda2647d421073","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 09:23:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4ea2c2ee32fefce12320758030b56cdb6197e075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 09:24:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c25b7a23f4da546d2c82b276d184bc3ce2ac04cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 10:50:47.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f6db49d4360614de11ba0cfeddf4078a5d3f8ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 11:38:52.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"a85614c894fc7c9f76731b96b50c0f5ecf5b05e5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 13:02:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f5a884b4f4dbb0e846e17541d7d0a1708573d971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 15:22:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef2a3e5afbb8476f9592587a50e5b454 : SUCCESS in 14m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fbbc0713cd1945dc85af651a0d60bb64 : SUCCESS in 1h 04m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1c0c6c9a78494d45a537d3947f7c7257 : SUCCESS in 1h 15m 17s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/826883ecc900426eaa5773011b206bca : SUCCESS in 1h 13m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85581fa99b864aaa93099e32b68541df : SUCCESS in 1h 42m 16s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8d935c58aa7450b9fa74719d9b366c0 : SUCCESS in 2h 03m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a5f09b5de4a450b862b7126c51dafa4 : SUCCESS in 2h 13m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c258926d6dd4892b1b0560b36e66e2a : SUCCESS in 1h 57m 32s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eb3fff1451c4a0d954bc6067bb0a160 : SUCCESS in 1h 58m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfad8cec7cd44e86a8205a96743c4cb7 : FAILURE in 23m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e84f0f781f94925a71feb5bd61c274c : SUCCESS in 50m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487befebdd824d668f2d6c59bc89606b : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":4},{"id":"4c3253744d3c04f4d1f0e24cb844218382ebb8b7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-24 17:30:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c096bead6f18bf84d0937559dc856fa4e157b529","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-25 09:03:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"77d524cf2b05a22603efa6cab88a8a42b5b72a36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-25 19:38:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dd8af55e86925818c7b365560d908c006abd3e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-25 20:46:29.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"b3bdbdb920dcbeb0862427e4aa686b9f03e2ca5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-27 17:21:27.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c4dc1e8ffde785b5dcca9b72cb5b42eb2dac7c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 19:42:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a310260295ac44be8e19f320751cf704 : SUCCESS in 14m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/626998c76cca48d79424249d9fb45138 : SUCCESS in 1h 05m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/000f57d635a94662a646c1167ac0445f : SUCCESS in 1h 32m 46s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a2a3680a81945b3ac3d7677dd0c7c9d : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de26a000885d4aadb3cf0219cd974aa4 : SUCCESS in 1h 46m 32s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/680041e871824cae93899e069d8d9468 : SUCCESS in 2h 08m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47b1793cac80403081946168e7d282ae : SUCCESS in 2h 13m 34s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84febfe590c2405eb8e4785de2ee10fa : SUCCESS in 1h 55m 27s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e8faa02b2e349e69c795dedb7231b52 : SUCCESS in 2h 03m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/83bd3a36d55941a18978c72bba7b9e83 : FAILURE in 32m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63fa0c1b96ff42198f434c6a0961380d : SUCCESS in 48m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cdfef9f6f0e4c26b3415cfd38a32e11 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":5},{"id":"3c0b40e45f8b0a3d3a3f5d461a17be240dc53182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 13:09:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dcb64a76bb89f231f1e212bf8904529236db3cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 15:04:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e6887ffa0ce41a28bb64f00bf2ae85c : SUCCESS in 14m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/66518d933bbf46789ddf913530349f39 : SUCCESS in 49m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/72173f1106804222951968e87f7e79f8 : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/683d768c444a44aab0092c9f339bc072 : SUCCESS in 1h 29m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b73e41b694954e8584365bbdcd176e44 : SUCCESS in 1h 20m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c388c5c669484dbebcb17a334762f06a : SUCCESS in 1h 54m 38s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/726189c002024bf8b7497bed4505bdb9 : SUCCESS in 1h 36m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e99e307039c34f639d74348d75155492 : SUCCESS in 54m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3415af5fdb0a46f69ab24a0726ff434d : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":5},{"id":"5caafc0845f6daf6757727db9aa9da5ece7e6ddf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 15:04:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ab56b27345d9407393194af0a6831df9ae61be92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 15:06:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163010a079ed4b3e8c94abb61a2b90d3 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a2694d31479beb57336af2d7880db9722a4be582","revisions":{"a5194e5783b54fef6fcd9aca029ba7deef16d16e":{"kind":"REWORK","_number":1,"created":"2021-12-24 03:15:45.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/70/822870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/70/822870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/1"}}},"commit":{"parents":[{"commit":"7a91492c6a18eda1bbb21e717f572683ce837520","subject":"Merge \"Stop managing NFS mount destinations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7a91492c6a18eda1bbb21e717f572683ce837520"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:15:43.000000000","tz":-360},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a5194e5783b54fef6fcd9aca029ba7deef16d16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a5194e5783b54fef6fcd9aca029ba7deef16d16e"}]},"branch":"refs/heads/master"},"2982142c277c35079a5517c70add3fe9baaa139f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-24 09:23:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/822870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/70/822870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/2"}}},"commit":{"parents":[{"commit":"5f565cd53940d63b7b31fa7d233388ae5b393df3","subject":"Define _glance_available_stores in variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5f565cd53940d63b7b31fa7d233388ae5b393df3"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 09:23:44.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2982142c277c35079a5517c70add3fe9baaa139f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2982142c277c35079a5517c70add3fe9baaa139f"}]},"branch":"refs/heads/master","description":"Rebase"},"74ed3021711ee5c80faf02c8fc24497a45ec1597":{"kind":"REWORK","_number":3,"created":"2021-12-24 09:24:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/822870/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/70/822870/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/3"}}},"commit":{"parents":[{"commit":"5f565cd53940d63b7b31fa7d233388ae5b393df3","subject":"Define _glance_available_stores in variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5f565cd53940d63b7b31fa7d233388ae5b393df3"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 09:24:31.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/74ed3021711ee5c80faf02c8fc24497a45ec1597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/74ed3021711ee5c80faf02c8fc24497a45ec1597"}]},"branch":"refs/heads/master"},"213c1e21867fcff4097b1b0fde2125d3e3772360":{"kind":"NO_CHANGE","_number":4,"created":"2021-12-24 10:50:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/822870/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/70/822870/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/4"}}},"commit":{"parents":[{"commit":"07341279f30df2fd0b43d135aa4c5424e05179f2","subject":"Define _glance_available_stores in variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/07341279f30df2fd0b43d135aa4c5424e05179f2"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 10:50:47.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/213c1e21867fcff4097b1b0fde2125d3e3772360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/213c1e21867fcff4097b1b0fde2125d3e3772360"}]},"branch":"refs/heads/master","description":"Rebase"},"a2694d31479beb57336af2d7880db9722a4be582":{"kind":"NO_CHANGE","_number":5,"created":"2021-12-27 17:21:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/822870/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/70/822870/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/70/822870/5"}}},"commit":{"parents":[{"commit":"091f33c8da9e355898cd4ffb8e2569481b7d65ab","subject":"Define _glance_available_stores in variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/091f33c8da9e355898cd4ffb8e2569481b7d65ab"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-27 17:21:27.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2694d31479beb57336af2d7880db9722a4be582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2694d31479beb57336af2d7880db9722a4be582"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
