)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~822899","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I312fdede3c3bffa0c897588d33fcc5d5d6988ccd","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2021-12-28 14:56:29.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-28 13:05:22.000000000","reason":"Dmitriy Rabotyagov replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2021-12-25 19:35:29.000000000","reason":"Jonathan Rosser replied on the change"}},"hashtags":[],"change_id":"I312fdede3c3bffa0c897588d33fcc5d5d6988ccd","subject":"Define _glance_available_stores in variables","status":"MERGED","created":"2021-12-24 09:23:31.000000000","updated":"2021-12-28 14:57:55.000000000","submitted":"2021-12-28 14:56:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822899","meta_rev_id":"883912f2339fd7287f0c90b968d8a12a8b9302f8","_number":822899,"virtual_id_number":822899,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-28 14:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2021-12-27 19:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":2,"date":"2021-12-28 13:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":1,"date":"2021-12-28 13:05:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-24 11:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-24 20:44:26.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2021-12-25 09:03:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f70e3e75ee6da37f18673cc15cb1a5f9d6dea878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 09:23:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cc021e6e370aa9d6bf3e7b9112a828be100d274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 10:50:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ed076e18708062e73ccb5e29b92c3d2058115c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 11:38:51.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"39c2acdafac0adfc89371a0d20e7135393d949e5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 13:02:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8da399db36c225233115df1ed95c5317a37e60fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 15:03:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42c53ac1eddb4ab484ad96f14b4eeb14 : SUCCESS in 13m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f667441470ba4f819e8678846d842f7b : SUCCESS in 1h 02m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e0b18ce020e44c01b0417fc5e57878ed : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c25149b00ba44950b456d5c9da1a0b43 : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5bb6d91cd8542acaf7b2a2e6b370b56 : SUCCESS in 1h 00m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a2619d19df74b1892a59b11b92b9484 : SUCCESS in 1h 45m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9aec724007d46038a2b49ae34e753d2 : SUCCESS in 1h 56m 34s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75872553a3a44f57b9677c87737c8930 : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f49bccf59ea4e4d903afc4ce537f59b : SUCCESS in 1h 37m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5da5f0901ca54a579532b75491355ce2 : FAILURE in 20m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae4714174cf24f389791c76fe78f64ff : SUCCESS in 1h 07m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c263bdb8240640c18e0722ded629f35c : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"66c68551036453ec119f0dc546fef7e8478d08fe","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-24 20:44:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"68d945477c4feb9999d26a5d1533d5a40cd23116","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-25 19:35:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35684babad66473aa3b86d3b8d5ffbb490719bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-25 19:37:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ffdb62a865ad0477f2d0519a59e53eecfac0605f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-25 20:46:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d163d099bd6648db9c4a17c21e35fd4b : SUCCESS in 16m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ab9bbaf1e534430c937f241216a894da : SUCCESS in 1h 03m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c7bdce390173494bb3c0b522847870cf : SUCCESS in 1h 01m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc4ec4ea4d8440679d421eebc0c34d26 : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0132b918404e4892bee29952749b3fea : SUCCESS in 1h 05m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21ab40c87bc448fa8ec350ee1481fe63 : FAILURE in 37m 10s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/662a84815e1d40d1b9388a709b2946fc : FAILURE in 28m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/760b04caee21423388fb98a5a23234ca : SUCCESS in 49m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a885e0ca2824f6394a3f5efdd4a46fa : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7f4b6ecbc9ee56d3887abd35760a4974ee0a826a","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-26 00:58:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"626c04365c2326349fca8335bf04bf4afa017c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-26 03:27:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b49e348fea2743d59a37540cb53daaa5 : SUCCESS in 11m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7d0572e68e524ca9a4a3d81978014057 : SUCCESS in 47m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0928510a59074172ac10ed0993af4a71 : SUCCESS in 1h 11m 57s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cd7c375f358437d881151fa77fde9d9 : SUCCESS in 53m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ce806ee6a4e4f5ab5ccc7eadfe13b3d : SUCCESS in 1h 01m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3715f5366d34de99e0e6d2e48698d00 : SUCCESS in 2h 22m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a2ac125898c4cb18dd4bff1af0e889f : SUCCESS in 1h 41m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b046d0023ffc4935b14f1d25fcad5dcc : FAILURE in 28m 47s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b986b63774404b1e8118051097fb0475 : FAILURE in 30m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a9fd240366f4b8786e541ce6a8f8851 : FAILURE in 19m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b28a1f7f9484de6a1a8ee6c9ba345cf : SUCCESS in 38m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32d38f4f43c473296a238509230af2b : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0109499590350892c1367bc66ff40fd2f2701fef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-26 16:51:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f90c9d78e3f78cdace66a4f25dd505c263d5911e","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-27 14:22:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a86cbff461bb7e55348c640c2ea5a172d242f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 16:50:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce6996bf4cee40a8bc2e2e3ded714f6f : SUCCESS in 14m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e8881ce00ffb483d8730ac58f1fb9e5a : SUCCESS in 1h 12m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/207ec9b72ee04e309b2b02e3814dd374 : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/af223ae5def34c19bb0c6c3e20b189bc : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a09ce185db4408dafb108f2411f76a8 : SUCCESS in 1h 02m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9b28ab306354d64bc7f61febc696f80 : SUCCESS in 2h 03m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/440e25071d354949b5f66ae4e8b07db4 : SUCCESS in 2h 21m 57s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f592b3bc907f4bf3861093b49525664c : FAILURE in 38m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f0df696289f4ff9a94033ecf24ed70f : FAILURE in 40m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc14041e4624431faa887a7e0f3cdee7 : FAILURE in 25m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49a2e265141d48b391afd1b5b6d72c5b : SUCCESS in 53m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37be026c230145eebfe1b31290eefb85 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"019c88de28b20f73daf88ec8c66caade95eb376e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-27 16:53:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"609c957133994ad7243b877a3bb2afec53c6bbc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 19:27:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25e5482084be405ca9b29072bb85e3f6 : SUCCESS in 15m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3e37ccd305204bf3b7e32ff36b27865b : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a52c298c05ff47a988a36009e43a6470 : SUCCESS in 1h 11m 18s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc4c53e8b13046e982f86494d41fe421 : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d641a9144aaf40d3a7bd282b9f92fe6c : SUCCESS in 1h 10m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ef315e0b9914b049ca114e69d7059b4 : SUCCESS in 1h 58m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd4fb54b50fa455683ebcc13fb721df5 : SUCCESS in 1h 59m 30s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55b9bef49f694458a98dcd06e816c296 : SUCCESS in 2h 01m 28s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da24cef0ab7149109a738b96cd78e979 : SUCCESS in 2h 26m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c3f79a440734bcb9d00b559964cda8c : FAILURE in 22m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/483c9384916e4ab1b439dd637013208f : SUCCESS in 53m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c336b42eea4285a284ed280a5e8b91 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"036184245376addf44426ee449ace2408aef947c","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-12-27 19:57:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"640c999b84b960517f383d225c3a0bcc5f134008","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-28 13:05:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a5d2e6570b27f00d7d9c210162c7b7eb558e1be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 13:07:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd942914449b4de9fca12c68ca7e005dacd04bad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 14:56:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e04483794bd445d9b5667328728927ca : SUCCESS in 28m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7f9330f992914fa8b280f9c923e71122 : SUCCESS in 1h 01m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9319cc3a81e24235ada85e4b76d9584c : SUCCESS in 1h 29m 15s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c25e340166f49eeb3b5e5b51e9d9059 : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/652912f9e21b43f993cad3a580ddbdd7 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76833997253247e4bee69718dac90599 : SUCCESS in 1h 43m 53s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73499a8a443d46d89d0fd9bec3784955 : SUCCESS in 1h 47m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcfd5fe8fa584fba9a165dd3dbb6c50a : SUCCESS in 1h 37m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bb452aa34464b939699479755143717 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f587f7ba753638caba9282e5c2c5db82ec1ae488","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 14:56:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"883912f2339fd7287f0c90b968d8a12a8b9302f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 14:57:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ddcd0a2b024c87b8fd7c446598ee1e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"091f33c8da9e355898cd4ffb8e2569481b7d65ab","revisions":{"5f565cd53940d63b7b31fa7d233388ae5b393df3":{"kind":"REWORK","_number":1,"created":"2021-12-24 09:23:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/99/822899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/99/822899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/1"}}},"commit":{"parents":[{"commit":"7a91492c6a18eda1bbb21e717f572683ce837520","subject":"Merge \"Stop managing NFS mount destinations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7a91492c6a18eda1bbb21e717f572683ce837520"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-24 09:23:14.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-24 09:23:14.000000000","tz":120},"subject":"Define _glance_available_stores in variables","message":"Define _glance_available_stores in variables\n\nGlance available stores is usefull outsie of the template as well, so we\nmove defining variable from Jinja template only to ansible variables.\n\nChange-Id: I312fdede3c3bffa0c897588d33fcc5d5d6988ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5f565cd53940d63b7b31fa7d233388ae5b393df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5f565cd53940d63b7b31fa7d233388ae5b393df3"}]},"branch":"refs/heads/master"},"07341279f30df2fd0b43d135aa4c5424e05179f2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-24 10:50:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/99/822899/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/99/822899/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/2"}}},"commit":{"parents":[{"commit":"7a91492c6a18eda1bbb21e717f572683ce837520","subject":"Merge \"Stop managing NFS mount destinations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7a91492c6a18eda1bbb21e717f572683ce837520"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-24 09:23:14.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 10:50:38.000000000","tz":0},"subject":"Define _glance_available_stores in variables","message":"Define _glance_available_stores in variables\n\nGlance available stores is usefull outsie of the template as well, so we\nmove defining variable from Jinja template only to ansible variables.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/822901\nChange-Id: I312fdede3c3bffa0c897588d33fcc5d5d6988ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/07341279f30df2fd0b43d135aa4c5424e05179f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/07341279f30df2fd0b43d135aa4c5424e05179f2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"091f33c8da9e355898cd4ffb8e2569481b7d65ab":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-12-27 16:53:47.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/99/822899/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/99/822899/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/99/822899/3"}}},"commit":{"parents":[{"commit":"7a91492c6a18eda1bbb21e717f572683ce837520","subject":"Merge \"Stop managing NFS mount destinations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7a91492c6a18eda1bbb21e717f572683ce837520"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-12-24 09:23:14.000000000","tz":120},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-27 16:53:47.000000000","tz":0},"subject":"Define _glance_available_stores in variables","message":"Define _glance_available_stores in variables\n\nGlance available stores is usefull outsie of the template as well, so we\nmove defining variable from Jinja template only to ansible variables.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/822998\nChange-Id: I312fdede3c3bffa0c897588d33fcc5d5d6988ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/091f33c8da9e355898cd4ffb8e2569481b7d65ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/091f33c8da9e355898cd4ffb8e2569481b7d65ab"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
