)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~822944","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fxena~I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa","project":"openstack/openstack-ansible-os_glance","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-01-03 16:20:16.000000000","reason":"Change was submitted"},"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2022-01-03 13:43:13.000000000","reason":"James Denton replied on the change"}},"hashtags":[],"change_id":"I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa","subject":"Add boto3 module for s3 backend","status":"MERGED","created":"2021-12-28 17:34:12.000000000","updated":"2022-01-03 16:23:02.000000000","submitted":"2022-01-03 16:20:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822944","cherry_pick_of_change":822870,"cherry_pick_of_patch_set":5,"meta_rev_id":"d6f94d97229ffcf3b554863a627db27bf93996ab","_number":822944,"virtual_id_number":822944,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-03 16:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2022-01-03 13:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-30 22:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":1,"date":"2022-01-03 13:43:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-28 17:34:12.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2021-12-28 19:36:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-30 22:42:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"ff5dbe8b17d890e88e04d17fca7879ac80421b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-28 17:34:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ee2c796e15cfb725f4cdf3920c279b8ce6826df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-28 17:34:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e2b9465b9dddd95f69d1f052c3e5466342bc37b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:36:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c0d3d1854af47d9ab8b147ea6f55bb1 : SUCCESS in 12m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c64e1085bee941dab77e0e6301495199 : SUCCESS in 1h 38m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8442c040ef134aa996d10c3ed8420c09 : SUCCESS in 1h 08m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4058abebbcf24f6c9a75a3e80ab5ad02 : SUCCESS in 1h 20m 23s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c807f8cc21c445928f58d3f9fe3b8ad4 : SUCCESS in 1h 46m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/922b245a61d8462999ccc831b4934eb2 : SUCCESS in 1h 48m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/620fa561a6244482bd2badc48bae33c6 : SUCCESS in 1h 58m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2aa1a68303ab4c0295f43e49cff30161 : FAILURE in 1h 25m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac045fbe2bb94bb0a8fdd07410c2529c : SUCCESS in 1h 59m 25s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a3f81313e4b4f62a3eab1e18bd02544 : SUCCESS in 1h 50m 16s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46d4bbb1b9594f9b9fa4f8de1deb685d : FAILURE in 53m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fde7534fb6dc492d81fa24e4dff7567c : FAILURE in 33m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e2e1e19d51746e6bb0ebc73f6c23091 : SUCCESS in 56m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e84e77d3c3f4211b3d7a7596a36ed86 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3a0e75515015ebff5f65ea06c509febc8c631b03","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-29 10:01:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36a87e187eab8a54023ca23abb649e2b78d1eaa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 12:15:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e002d6f2085d4afd803153d55d78b836 : SUCCESS in 13m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c5731504c0b64aaf82ae045b2f8a5994 : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2e9c0c064ec8486081f39edd76d04f43 : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ffdf338b89e0461ea34d56c438039062 : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6fa42ddd9bba40eeb87aa4c1872cbdc7 : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f0a28a3771840e787d8bcb1ea529586 : SUCCESS in 1h 25m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55800935c35b48e0aaff2fb59cf6bcc6 : SUCCESS in 1h 19m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa247306af09414db7c183df076117d7 : SUCCESS in 1h 45m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8b5e863adf94efa85a9e216cf508af4 : SUCCESS in 1h 41m 52s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ef5b8a247a4787bf3544b230980e84 : SUCCESS in 1h 58m 00s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b222ea732fc145e0b0f40eae5e77be0b : SUCCESS in 2h 10m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8433effd178d46fd9674398f085d9371 : FAILURE in 29m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f937cae75bab4e949a4447c1e20e816c : FAILURE in 39m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f9bf8578f347f89eaca88d0d7e9ae7 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"71215e67c7162e21c6d91c8803541990d3143011","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-30 05:50:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f542b4487e1a4ac836af3953490902126541ad23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 08:10:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae11f5a41b7747379905dd3d2a87f588 : SUCCESS in 14m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/445e34641a6144d4850743437efb6f3f : SUCCESS in 1h 36m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/630e8f8b39dd4d1bbab039212a1c5ca5 : SUCCESS in 1h 04m 27s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/208cad8d685747b9af6e01d9945dcbd7 : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7654ca69533452d8f779f9cf14e6bbd : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40457905821447b0984ca96c1f7c308c : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ff3d75df4a44b8caba1a333d8b373af : SUCCESS in 1h 19m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e138b10d17f438eaf3cc7af9d22b31c : SUCCESS in 1h 43m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74ee2f8086d34fb4bb65e619323de340 : SUCCESS in 1h 54m 40s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e1ee7d493344b958c6361bd828d3eaf : SUCCESS in 1h 44m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec6c22922b234f2080be0ca475fa16a1 : SUCCESS in 2h 16m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/63aac8b1d70848149e2f0fe81abdb2f7 : FAILURE in 31m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcd1e11d55fc43048bae6cb49b71ac72 : SUCCESS in 1h 00m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86a7c11b22ee4b14bde26a03e24b0ceb : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"41923f66b6185bfbe6c7577eefcb95c7a962526a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-30 22:42:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"573c692934cfa099cc969533b9ceaf4240258257","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-01-03 13:43:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ebd598edd7d4256af6372998c02e255d506da2b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 13:48:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3bfb6ca23869fb1c612f9ca5a56fe5f9e028e889","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 16:20:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20f3d03ce1c84a90910b1642c461e3a0 : SUCCESS in 16m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/27055eac7e924ccda8d394b187b2253d : SUCCESS in 1h 43m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dc3da698061444ea82ec136fed844608 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fced05d4fd704a20b81f6c7a5116f058 : SUCCESS in 1h 30m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2d0db38ff884538a8223bbd3896c30e : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bf57ec2953e4f40995809d81b8dda8e : SUCCESS in 1h 45m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eab7d6b358542e59451e46bc2101c38 : SUCCESS in 1h 28m 58s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e739c455a814bb2be569737951e6f9b : SUCCESS in 2h 14m 50s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/965e1c965bc94bc2a7752409968dabf8 : SUCCESS in 2h 26m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af0e0d2431be4a529626a8228c23b326 : SUCCESS in 1h 05m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83a13740bb844208ffadced77fea218 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e50769795774889abd06e406be7e475d3fb8f0ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 16:20:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d6f94d97229ffcf3b554863a627db27bf93996ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 16:23:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc9ffb9e33348b8a21998d392edadc3 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5305adfcef6956036f90262e5cf42cc6fd59be4","revisions":{"32889a565ac2d3b4c6ecc6e941b2357f071b1c54":{"kind":"REWORK","_number":1,"created":"2021-12-28 17:34:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/822944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/44/822944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/1"}}},"commit":{"parents":[{"commit":"f47a5475d1e46d4b4fd6f8f1741406cc2660d0d0","subject":"Stop managing NFS mount destinations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f47a5475d1e46d4b4fd6f8f1741406cc2660d0d0"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-28 17:34:12.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n(cherry picked from commit a2694d31479beb57336af2d7880db9722a4be582)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/32889a565ac2d3b4c6ecc6e941b2357f071b1c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/32889a565ac2d3b4c6ecc6e941b2357f071b1c54"}]},"branch":"refs/heads/stable/xena"},"b5305adfcef6956036f90262e5cf42cc6fd59be4":{"kind":"REWORK","_number":2,"created":"2021-12-28 17:34:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/44/822944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/44/822944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/44/822944/2"}}},"commit":{"parents":[{"commit":"f47a5475d1e46d4b4fd6f8f1741406cc2660d0d0","subject":"Stop managing NFS mount destinations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f47a5475d1e46d4b4fd6f8f1741406cc2660d0d0"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-28 17:34:28.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n(cherry picked from commit a2694d31479beb57336af2d7880db9722a4be582)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b5305adfcef6956036f90262e5cf42cc6fd59be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b5305adfcef6956036f90262e5cf42cc6fd59be4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
