)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~822946","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fvictoria~I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa","project":"openstack/openstack-ansible-os_glance","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-01-17 10:57:01.000000000","reason":"Change was submitted"},"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2022-01-03 13:46:08.000000000","reason":"James Denton replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-01-17 10:57:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa","subject":"Add boto3 module for s3 backend","status":"MERGED","created":"2021-12-28 17:34:49.000000000","updated":"2022-01-17 11:01:13.000000000","submitted":"2022-01-17 10:57:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822946","cherry_pick_of_change":822944,"cherry_pick_of_patch_set":2,"meta_rev_id":"a6bbe9af1bc1665bcfcd44dfcd73c20ca8ec480e","_number":822946,"virtual_id_number":822946,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-17 10:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-01-06 13:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-15 08:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-01-17 08:30:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-15 08:58:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-28 17:34:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2021-12-28 19:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-29 12:07:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-01-15 08:58:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-01-17 08:30:15.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"444f2f30ffe309622c664e28e9a14792a94433aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-28 17:34:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"fc554bd744e2b91ff78e32d68b1c54621430fddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:35:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/452c5c3c2b424ff3a78b209cfcc29b8e : FAILURE in 10m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/08f409e3b6da4b259c65bf909f4cc14c : SUCCESS in 1h 19m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/32e93f78c9534df09c498fe311dc268d : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58e3b02765ec4e3ebefb45bd2b7041c3 : SUCCESS in 1h 36m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b061ceff31942d7bde105a1a0dce724 : SUCCESS in 1h 07m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ad1ae4ea456e4e659cd159e2216e5645 : SUCCESS in 1h 52m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6cc24ec71ce4d73a58006f510382e18 : SUCCESS in 1h 35m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cb1566e103c145478985e43d82de2832 : SUCCESS in 1h 03m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff2da2119b5d408b8fd4958ff094707f : SUCCESS in 1h 19m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ace73034a54499694dab157dcd597c6 : SUCCESS in 5m 27s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/f2cc391b46814edfad11b90f78bffaf0 : FAILURE in 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6ae8e6c29a8b054f9674e4c9c3a17b8a32061f4a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-29 12:07:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f6a7e12b39175c5c9cb5ce11e5d894a92eb2f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 14:10:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c6f69d82f05243a090dbdb8527059df6 : SUCCESS in 15m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/28a7ff55a98a4895acc9d36a79b99ef4 : SUCCESS in 1h 04m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b5f46c824b21411bae921b7ca8a10aa5 : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d3743fdea1d4ecca2093474bac9e159 : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2911d19b43994386ac514a4d00a00a6e : SUCCESS in 1h 00m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e33718177f994410ba75679b4bc0fa9e : SUCCESS in 1h 57m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e5a7ff5da714bdfb145125a9918d492 : SUCCESS in 1h 23m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c077168a30d94d62adee09db85a685bf : SUCCESS in 57m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a75ea2ba34d84e238e491e9090b7d10c : SUCCESS in 1h 04m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29fb2ed906484555aad8a74c9897959b : SUCCESS in 6m 48s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/be9cc3b482944810aba83c927cad9bd1 : FAILURE in 37m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9b4725e474359238067fa1da0e49a6653fce4f60","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-30 05:50:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"055fe90166cd283ec9090fcc8c4db5c2d7e221a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 07:33:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0feff259881445209863ae6d1f0db8a5 : SUCCESS in 11m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6b1fc0d939e74397b5010f79cbb599bf : SUCCESS in 1h 17m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8ae6bc5f9161428a848818e1a1821834 : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c590b8e97d249b2ac24568d98f93f10 : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e32eb6cf3cd94e5da2d8d45446001793 : SUCCESS in 1h 08m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bb2a7ff6593d46b5a09a793f4ae5bc2a : SUCCESS in 1h 40m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5888f5b31854de1842b6436910f94da : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8694ad6796214a02b4e365800e51efe9 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7429da23dc034e9a877b5cc7f8395210 : SUCCESS in 58m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bccb165bd39c4970a2c4019412417a9e : SUCCESS in 6m 14s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/3e4ecfa044d34305a6a9f6be38784cb7 : FAILURE in 47m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0b1bbbee14afe34c3118697e1aae9af3eba779e4","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-01-03 13:46:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5261f84bec5e0ff356d5ab616e510fd95f02cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 15:46:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2314dda8296b4fbdbdbd6bc5a375f658 : SUCCESS in 13m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4f5a36fdaccd46c89c0ad118b15f6b3b : SUCCESS in 1h 14m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6f86b3fd337c4a06947aed70d70d04dc : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e58ed6807ea4e208798ef0c3c0e1945 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fe66b3811b1434b93ffce58feeda691 : SUCCESS in 1h 17m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a285b5a22ec64768903973812345ce5d : SUCCESS in 1h 54m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0adac79d53c149f48e678bd415cdaf31 : SUCCESS in 1h 41m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e7c98d27c9bc4fa998d5db98323fae67 : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39041020b13043d6acbf0558e212dd03 : SUCCESS in 1h 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1898c9da17d14744846d7ec186533642 : SUCCESS in 6m 44s\n- openstack-ansible-nfs_glance https://zuul.opendev.org/t/openstack/build/3082a37726e04d27ac7251b966eed86a : FAILURE in 42m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4711329010bd89b877dd08e25fb83991234f6429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-06 10:08:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"595637498951c155f14701677a95b2d73aac9ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:32:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d9ea632fb9f843bebd50dd2a7d24c3e8 : SUCCESS in 12m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7e080f05e8544dfda84726551375890e : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7907a3fb7a904eb190527c5677b89c18 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06c4a3f58a3b4532ba04e3ba5fd61e1e : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25c12109eb504008a57367a3caa0ce7b : SUCCESS in 1h 12m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bf30af8918b740d9aedebd873fe102a9 : SUCCESS in 1h 56m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33bd3d043c6847ccb97ebfb0c2642a7c : SUCCESS in 1h 35m 41s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9b8c80e0c734d28aa20638658c30329 : SUCCESS in 1h 50m 55s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd8505141b1641b19278babbe3cf7428 : SUCCESS in 2h 18m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e250b5ae8cac49de9247ec2a3a567b53 : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eade3dc17e2d4ae8a8bcef93423f3d26 : SUCCESS in 1h 04m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc134cb75a443f0ade42300a8f79517 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"830af31e44ab5045480ad690f62d2e46a436d0a7","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-01-06 13:58:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"068ef64b4c67fdcf768526b5cdbcbf22a9154d9f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-15 08:58:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c80533aeb1dca312f91a5dbfc640e6b05edc49ae","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-17 08:30:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"44557e5642c3e2b5ea8254962be1f405e6ea80aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 08:32:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"34d5f864e4328428a0bf8b3053a95d15df1529df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 10:57:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fcf93400151340cea6b645e4b5e001af : SUCCESS in 11m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7c3c992782b2420abdda3e7fd1bed61f : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5933e4e3a0b34328adabc0ad7a81a174 : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a125a9b24c11405eba6ce303a264bad6 : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae3fb03cf10843cf97a5f7884d7b0dff : SUCCESS in 1h 11m 36s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4637664f75f949e6a3ad4808978c09e8 : SUCCESS in 1h 48m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a4dd1373ec046e1a5b230b858d588eb : SUCCESS in 2h 20m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f9bc795c04274e7882c4cc53578ec1a5 : SUCCESS in 55m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31f79c2ca25943b7aa7c88cbb2c16fde : SUCCESS in 1h 04m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eda219ce9d545dda1e04f02052e4025 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"9703545192e265780285894304f51313abb320c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 10:57:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a6bbe9af1bc1665bcfcd44dfcd73c20ca8ec480e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 11:01:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc50dbfbb6b64332a0e8f339a651a1a6 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"084b6b4d9dfca5002e48d8d66b5d9a5332883739","revisions":{"b378c69b77e550f7951e8cd7ef8ee8eb441e1d07":{"kind":"REWORK","_number":1,"created":"2021-12-28 17:34:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/822946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/46/822946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/1"}}},"commit":{"parents":[{"commit":"59f398fb176f2c45d6338933693f344e6e5e7ce2","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/59f398fb176f2c45d6338933693f344e6e5e7ce2"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-28 17:34:49.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n(cherry picked from commit a2694d31479beb57336af2d7880db9722a4be582)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b378c69b77e550f7951e8cd7ef8ee8eb441e1d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b378c69b77e550f7951e8cd7ef8ee8eb441e1d07"}]},"branch":"refs/heads/stable/victoria"},"084b6b4d9dfca5002e48d8d66b5d9a5332883739":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-06 10:08:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/46/822946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/46/822946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/46/822946/2"}}},"commit":{"parents":[{"commit":"ce5fde12e60246a35574f8d408dc2386e3b701cd","subject":"Replace NFS test with integrated one","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ce5fde12e60246a35574f8d408dc2386e3b701cd"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2021-12-24 03:13:19.000000000","tz":-360},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-06 10:08:59.000000000","tz":0},"subject":"Add boto3 module for s3 backend","message":"Add boto3 module for s3 backend\n\nThis patch adds the boto3 python module, which is a dependency\nfor the s3 glance backend.\n\nCloses-Bug: #1955683\nChange-Id: I5f5a921d8a08d5dfb09e2bca71d49c85115a60aa\n(cherry picked from commit a2694d31479beb57336af2d7880db9722a4be582)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/084b6b4d9dfca5002e48d8d66b5d9a5332883739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/084b6b4d9dfca5002e48d8d66b5d9a5332883739"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
