)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~824397","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ica9144f21cd675112d07008a387f65f8e468d378","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/common_tasks","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-01-28 21:21:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica9144f21cd675112d07008a387f65f8e468d378","subject":"Use common service setup tasks from a collection rather than in-role","status":"MERGED","created":"2022-01-12 11:41:53.000000000","updated":"2022-01-28 21:23:39.000000000","submitted":"2022-01-28 21:21:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"824397-osa/common_tasks","meta_rev_id":"d5db7d8361ced2da361f63c73cb16625286d1642","_number":824397,"virtual_id_number":824397,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 21:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-01-14 08:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-01-13 11:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-01-14 08:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 14:53:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 16:45:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-01-13 10:59:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-01-13 11:28:03.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2022-01-14 08:46:59.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"554f408afa08f095e51dcf902b61d200fbd608d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 11:41:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d69cb4bdcf6ecf13a96263d7bdfe4173a52a1a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 14:53:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caf8c8a7f6ae4ad285d8d7040abffe43 : SUCCESS in 20m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5ac54fd1cee6452ebea95105c21b6d5a : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/92260e1543924acdb750dfbcaf051787 : FAILURE in 1h 23m 40s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e261f22f383b4ce08a8abdcd3ea85d0d : SUCCESS in 1h 39m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2614bd5cba44404fb11e83afc590a834 : SUCCESS in 1h 39m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc218dfdffa84f55918c350c5e30d239 : SUCCESS in 2h 13m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ade69436dcf43c190f4a01082827161 : SUCCESS in 2h 22m 26s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f0c031ad10341f785c79f28207aafff : SUCCESS in 2h 20m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb6e385259a84440b7a6b930825d6516 : SUCCESS in 2h 33m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbc4bd97542e4373a3e4f1892216cddd : FAILURE in 33m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b6d01021f5648d2b0d8a497ca438fa3 : SUCCESS in 1h 03m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdaff92da9be44ff821365f7e3e52cbd : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2ceeec0bfa4180846a9ee52ab920a24d10dca167","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 15:59:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64bba7563c1dd0307a0f2455bc4254c99fcebb42","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 16:13:09.000000000","message":"Topic set to osa/common_tasks","accounts_in_message":[],"_revision_number":1},{"id":"5e32dd49968ca8ef1d1201c2f8a31f0ee8c800f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 16:44:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb3d1db3b02b784486a1e2ae568a890547099417","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 16:45:45.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0c279df8d912bc5e3ca340f31f5890aac05a8945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 16:47:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"334f5d170cceb9851b61d121309e006fc9040156","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 17:13:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0a2cec942b19d4abe67be6e256bc5534a17e6642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 17:13:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"927cdfdcacdd3fcbd51a3f449e2fe9f6405e3da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 19:56:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e5f0eb92c91408799e2423c3b24c133 : SUCCESS in 23m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/021e48730cea421798404155d8e1997e : SUCCESS in 1h 22m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dcc0bb773b4a430c8e160efade4d02b0 : SUCCESS in 1h 35m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b523f42965544b5494588236d724e72c : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10578f194fe84239b2fca39a1ed2495b : SUCCESS in 1h 36m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/514ed5e0501840f7999f51a8756a5844 : SUCCESS in 2h 08m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e07524eeea845bf8ac1da48b993188f : SUCCESS in 2h 22m 47s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d5c0b22d1c54730842186c220d8ce34 : SUCCESS in 2h 22m 41s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db8c2fd64e5f4bc18d21b03d9edccb7e : SUCCESS in 2h 35m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ec0f0d10d9d46868590f22997d20f88 : FAILURE in 30m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed7629cf11c4d77a6c6a73adc128d6c : SUCCESS in 1h 07m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c944ed3bdde45ff842353a0dea8231e : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":4},{"id":"1e4cbb97c7c3dee91646d406b4c1f3747a1542f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-13 10:59:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a59ede8888e4a58b64d529406b46d75053fae242","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-13 11:28:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"288398af1b69446838a43b4aebf380a3358e6cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-13 11:30:16.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"80254cf66db141fdb2fef2f6f90d5bb5f62cef84","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-13 11:30:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fd950abc4f6f47cd35996f6cbdec04b514a59b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 14:22:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f92f94ef60da450ab729a6941ca44c55 : SUCCESS in 19m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/787bb7f993974e90a1c4860b5ffa2a24 : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ecabd59d2c524195b13994e4ce0b6f9f : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/48dd24411c264ddb8d31edca02d8aa2b : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/485b68d0bb0b4692a95d067093df0ffc : SUCCESS in 1h 31m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/812b20f31b61440bb4146b5acd97b0a2 : SUCCESS in 2h 06m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b242804610be456e87eeea05b51ad948 : SUCCESS in 2h 19m 46s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a002e63098e04200a55484941cc1e8f3 : SUCCESS in 2h 06m 51s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90c38f6e994c43e7bab929f4b0e1d66b : SUCCESS in 2h 22m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3290d51d4c64e6eab0244ceec5fd76e : FAILURE in 30m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb553c750241489e9a3e77947698ce59 : SUCCESS in 1h 00m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab6f1a455414b73865761631bc27678 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":5},{"id":"0e7207f3489b19c4983532466d05600fd3d3a8d4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-14 08:46:59.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3b53958a13de49503481e9fafcf5227f5ff3c883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 08:49:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9738ee2e19a82461e239636ebfa51fa386ee62e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:25:52.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4716487785a4fadb530a5d403cf5d61 : SUCCESS in 16m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/32d44fa653884f2d93bbdb234e000651 : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f55e2531a3d4d0b9a5e20ffc5e1c681 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3f237c3f2c14829bedf92b289d693b3 : FAILURE in 1h 17m 51s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/147be265b2cc4f09bb0539bf6560b005 : SUCCESS in 2h 12m 41s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19f08c405fb04c3491a008478792fe61 : SUCCESS in 2h 34m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6fad38789924d8682ae6e0ed2755351 : SUCCESS in 1h 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f67f688fa7944a8992e5477988735a1 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":5},{"id":"b2221b2bcddea4d1736624860ef8703225f1e5fc","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-28 15:44:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74814448b2f0ede1b2cfd6cf76b345b01cb69b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 18:22:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c78630358544ad0bb5899d1d5e5a786 : SUCCESS in 15m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/44ee810f34b94e4b9b60073aadcf80fb : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cb3849434bd41f6a1826da03d2b229c : SUCCESS in 1h 25m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f0f1829be1d4fdf840481e20e16a854 : SUCCESS in 1h 20m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2eceb2cc2fe54071bc5378efab49e0e2 : SUCCESS in 1h 48m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3a6ef5eb21c4be89dc2a72c59068826 : SUCCESS in 2h 11m 47s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f312e83912d94bd18995ace2d17160bd : SUCCESS in 2h 12m 31s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9939914fd9bd4e1fa7d0452557484c83 : SUCCESS in 2h 23m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef2845b7d69845f7959db19963ee2d63 : FAILURE in 30m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64e63c28cce2499c93e215f28fbd1164 : SUCCESS in 1h 07m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20424a966ed14488a83b42fc93c460d4 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":5},{"id":"d4a5d72bbe7ceeb42e7d7ea40c78fd1744f90fca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 18:42:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7c37785ab701c405cea817d6dbbf9fd9b662e118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 21:21:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a12a0b44a02d440b8f44b62ab31a9a07 : SUCCESS in 32m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b0ba97a94e03466f98580eb530cd2dd2 : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdbd7a9418074a32828ebcee283621d7 : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afacd989c00346cc9e2521b400b69478 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33137c46c87b4df5ad319f2f20626a14 : SUCCESS in 2h 07m 29s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10bf7a10da14409285806618b3333f68 : SUCCESS in 2h 37m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e58c2cfd6087437ea6b6b654830d5cd5 : SUCCESS in 1h 09m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cae8b61bcf64c50be6dfd0a780b2857 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":5},{"id":"10c5233e692f03180468dc2bca431ed4f287447a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 21:21:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d5db7d8361ced2da361f63c73cb16625286d1642","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 21:23:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20ae869cdfb412fb5323215ab8a2f4e : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9f54a5a634bdf9c4fe77b70cee73cd0a7f8565a6","revisions":{"2a74ee91d3a47153a687addb99f06d6a4f578a86":{"kind":"REWORK","_number":1,"created":"2022-01-12 11:41:53.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/97/824397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/97/824397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/1"}}},"commit":{"parents":[{"commit":"a2694d31479beb57336af2d7880db9722a4be582","subject":"Add boto3 module for s3 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2694d31479beb57336af2d7880db9722a4be582"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 11:41:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 11:41:29.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Ica9144f21cd675112d07008a387f65f8e468d378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2a74ee91d3a47153a687addb99f06d6a4f578a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2a74ee91d3a47153a687addb99f06d6a4f578a86"}]},"branch":"refs/heads/master"},"7dc604a977d0f9e29d26cfec823377b298c1033b":{"kind":"REWORK","_number":2,"created":"2022-01-12 16:44:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/97/824397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/97/824397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/2"}}},"commit":{"parents":[{"commit":"a2694d31479beb57336af2d7880db9722a4be582","subject":"Add boto3 module for s3 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2694d31479beb57336af2d7880db9722a4be582"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 11:41:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 16:44:07.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Ica9144f21cd675112d07008a387f65f8e468d378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7dc604a977d0f9e29d26cfec823377b298c1033b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7dc604a977d0f9e29d26cfec823377b298c1033b"}]},"branch":"refs/heads/master"},"f5d2e0e2ee2d14d617060c8871e1ca3db9b7d332":{"kind":"REWORK","_number":3,"created":"2022-01-12 16:47:39.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/97/824397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/97/824397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/3"}}},"commit":{"parents":[{"commit":"a2694d31479beb57336af2d7880db9722a4be582","subject":"Add boto3 module for s3 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2694d31479beb57336af2d7880db9722a4be582"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 11:41:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 16:47:39.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Ica9144f21cd675112d07008a387f65f8e468d378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f5d2e0e2ee2d14d617060c8871e1ca3db9b7d332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f5d2e0e2ee2d14d617060c8871e1ca3db9b7d332"}]},"branch":"refs/heads/master"},"aa05f8e351a30bfccf18c5a49f0277d63c50174a":{"kind":"REWORK","_number":4,"created":"2022-01-12 17:13:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/97/824397/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/97/824397/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/4"}}},"commit":{"parents":[{"commit":"a2694d31479beb57336af2d7880db9722a4be582","subject":"Add boto3 module for s3 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2694d31479beb57336af2d7880db9722a4be582"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 11:41:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 17:13:34.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Ica9144f21cd675112d07008a387f65f8e468d378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/aa05f8e351a30bfccf18c5a49f0277d63c50174a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/aa05f8e351a30bfccf18c5a49f0277d63c50174a"}]},"branch":"refs/heads/master"},"9f54a5a634bdf9c4fe77b70cee73cd0a7f8565a6":{"kind":"REWORK","_number":5,"created":"2022-01-13 11:30:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/97/824397/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/97/824397/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/97/824397/5"}}},"commit":{"parents":[{"commit":"a2694d31479beb57336af2d7880db9722a4be582","subject":"Add boto3 module for s3 backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a2694d31479beb57336af2d7880db9722a4be582"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 11:41:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-13 11:30:16.000000000","tz":0},"subject":"Use common service setup tasks from a collection rather than in-role","message":"Use common service setup tasks from a collection rather than in-role\n\nChange-Id: Ica9144f21cd675112d07008a387f65f8e468d378\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/9f54a5a634bdf9c4fe77b70cee73cd0a7f8565a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/9f54a5a634bdf9c4fe77b70cee73cd0a7f8565a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
