)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~835898","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Id99bbaee41bcdb103f719768839cd921163095ca","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"workaround-setuptools-autodiscovery","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-03-31 15:12:16.000000000","reason":"Change was abandoned"},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2022-03-30 21:14:42.000000000","reason":"Neil Hanlon replied on the change"}},"hashtags":[],"change_id":"Id99bbaee41bcdb103f719768839cd921163095ca","subject":"Disable setuptools auto discovery","status":"ABANDONED","created":"2022-03-30 20:18:44.000000000","updated":"2022-03-31 15:12:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d0b03960d136c8c3fc058717cea3891027801e33","_number":835898,"virtual_id_number":835898,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-30 20:41:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-03-31 13:27:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2022-03-30 21:14:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 20:18:44.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2022-03-30 20:41:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 13:27:44.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"ed6432051da0555498625ff8fbc0ca8ffbfcb607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-30 20:18:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaae646c79a55ca0e512f2f445e0f234a2493d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 20:41:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed21189aa8a141a1942f259e22b45ef9 : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8893cedd144653830eafdf785be574 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8cc81ca3e655bccf5e903253a1740eb9c9dd8400","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-03-30 21:14:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a6f43b5292dacc8dce993b6d1093c2af3fff4627","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-31 12:10:24.000000000","message":"Topic set to workaround-setuptools-autodiscovery","accounts_in_message":[],"_revision_number":1},{"id":"8e8a041d905eec9b99a3b496bf9a3e2254091009","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-31 12:51:10.000000000","message":"Abandoned\n\ncleanup instead","accounts_in_message":[],"_revision_number":1},{"id":"33f755fffe4d39d36761d173a8ece773474a6330","tag":"autogenerated:gerrit:restore","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-31 13:27:38.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"58d47dcf1eedcd1d3991c18bc0b75dd7eeba0e44","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-31 13:27:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"caf71fd2f8e737c4a343f7f19366c8b0299e8d83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 13:54:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bec6f2068b04c1e938e14bb57492965 : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8193ce4dd9ce446798dcb271b3af3d70 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d0b03960d136c8c3fc058717cea3891027801e33","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-31 15:12:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b35c0bf7e47132972239b6d322e10449ea7bcee","revisions":{"5b35c0bf7e47132972239b6d322e10449ea7bcee":{"kind":"REWORK","_number":1,"created":"2022-03-30 20:18:44.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/98/835898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/98/835898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/835898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/835898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/835898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/835898/1"}}},"commit":{"parents":[{"commit":"cab9cb90f5d48919388ab3d2bda4ea50f01d1c7f","subject":"Remove legacy policy.json cleanup handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cab9cb90f5d48919388ab3d2bda4ea50f01d1c7f"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-03-30 20:18:39.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-03-30 20:18:43.000000000","tz":-240},"subject":"Disable setuptools auto discovery","message":"Disable setuptools auto discovery\n\nJobs are broken due to a release of setuptools 61.0 introducing\nnon-backwards compatible breaking changes [1]\n\nSetting py_modules to an empty string here will disable autodiscovery\nand should permit jobs to run as before.\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nSigned-off-by: Neil Hanlon \u003cneil@shrug.pw\u003e\nChange-Id: Id99bbaee41bcdb103f719768839cd921163095ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5b35c0bf7e47132972239b6d322e10449ea7bcee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5b35c0bf7e47132972239b6d322e10449ea7bcee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
