)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~837550","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ic0d91a3a873b4253255beac79becf01b4a304695","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-04-25 22:47:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic0d91a3a873b4253255beac79becf01b4a304695","subject":"Use systemd_mount native syntax for mounts","status":"MERGED","created":"2022-04-12 10:56:49.000000000","updated":"2022-04-25 22:50:11.000000000","submitted":"2022-04-25 22:47:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837550","meta_rev_id":"24cfbd7fd98e74970104f70de469cf285b09a495","_number":837550,"virtual_id_number":837550,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-25 22:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-04-20 13:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-19 16:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-04-20 13:43:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 13:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 16:15:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-04-20 13:43:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"d1f4c80356e6a21f86e4270cf7d5db1a7d01b12f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-12 10:56:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb0c0efbd955e36c6b034812a57e7f9ac7b68f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:19:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a9c7ff9499344a7b9f419164d66af9d : SUCCESS in 14m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c5eac7c46fd14c45bb644688c59070b0 : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/600173a5c7dd47dd94ed1ff3aaa00b0d : SUCCESS in 1h 15m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/304c16a640d74f7083f672cdd9b77a2a : SUCCESS in 1h 22m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7fb99c129c124753b532238aeb669237 : SUCCESS in 1h 30m 50s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc019ba389114b979ebd46e30164dca9 : SUCCESS in 1h 41m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32982ab9580e4bd698ccabd3b900ec8d : SUCCESS in 2h 09m 00s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f140e5b05db452fb3d611fb3f41bef9 : SUCCESS in 2h 10m 37s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/381191c07bf8490d95e32c825fa95f69 : SUCCESS in 2h 01m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/16d6681d1fff4468a64dc52295e11915 : FAILURE in 31m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9feec62155444a498ef0c4252042393 : SUCCESS in 52m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae089ed44714c65aa198e5ba9299b19 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a38834f298a5460d991afd0b09807b7d : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"19d18dc20da265a824482f9e692a1e55df9879cf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-19 16:15:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d88d0ac08c1392b1bc9b46f9443f75f86bfe850","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-04-20 13:43:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed87cfe518c2d2661679ed9056ffe2502767906c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 13:45:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01dee8b80b768c9e1658aecf2a0276a09f774313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:06:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcdf3528728b4babbce7ada4c1152d90 : SUCCESS in 14m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/55828d0dd8794d4580d074437de34a1b : SUCCESS in 1h 13m 13s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/259b81aee51f4202a62acd7c0562be62 : RETRY_LIMIT in 14m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bb2a4c13f1341f49f2f6a351f640862 : SUCCESS in 1h 21m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c457f99fea05465eb0cf073b33025bbb : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd08c47787742d392bb3c7c5245fbe3 : SUCCESS in 2h 15m 48s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3774b033b4d440384763a9ae5061ffc : FAILURE in 47m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f3b5fd2049a4e68ab081d84607529ad : SUCCESS in 52m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2705cfd7204b3384075c329122eecc : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f43c2240826d41048f07b3bd56b4935d : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8633157d2afab273d210babbcf712a16e73cb173","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 21:48:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5acf49759c29e1ba8e86d75d45ed43f6b9279d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 00:19:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05b60cb699fe44468bd520d0a4e4d5a1 : SUCCESS in 14m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8011a86d4ab6460db50929b911eddd36 : SUCCESS in 1h 01m 00s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/70b3d90b76b045419dcdad1e8eb2b5fc : SUCCESS in 1h 19m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97824240118248668b6dacb9cba64b57 : SUCCESS in 1h 12m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d1f352c5d96c43039f69e6d38a06bb49 : SUCCESS in 1h 15m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/135b42050bb045ec9900979a83d5fcd9 : FAILURE in 17m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47d39cb92e44414ca6f10285da78223e : FAILURE in 13m 49s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/307cf3afa74246958cbf0b2d05ac79de : SUCCESS in 2h 19m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d029e2af0d6643b38ba3d585d7ecb53a : FAILURE in 38m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2431117083944694a67fe81e88126647 : FAILURE in 34m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/794246c1cd7e4b9fb84507380908093a : SUCCESS in 52m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11293d0b96844a6be86c068f23aaebb : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb47e37fd911434c8bb70cef771ffadc : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c3c3c4a30aa614ee8be52792caa58c2818ea5a29","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-23 11:24:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab810c58d7cf7f4be688e8fe2ccc55f808d3334a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-23 13:06:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfcd36b37b8c4991ad2d1ff402736cd1 : SUCCESS in 14m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4e3a64c5c5a349d69f066b7a03dc867e : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7db2008c8cf48bbb5c2370904316107 : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cb3c9094e3440dca25238f5a5d493c4 : SUCCESS in 1h 09m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8787d205e0874f4c8bc80ae573e7b3fb : SUCCESS in 1h 27m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/958442e6c5a346d5918495bdfb1d9f71 : SUCCESS in 1h 38m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5112235be1348a2bf22e7cf10eef283 : SUCCESS in 1h 37m 49s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8997ad48465e45338ac2b920db76130c : FAILURE in 46m 49s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52decc8eace14649ac1a20658497f3bb : FAILURE in 43m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fc5f189d43247609f276c8f00044d1e : FAILURE in 26m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407067dd719340a4a9b3291edea7d384 : SUCCESS in 47m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e410833e75492490cfc82b8779222b : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2b759442ea2494aa8c6c1add917dc1c : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2a82f4300c91cf76f48a4f96523dd9826db84294","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-25 17:59:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f6ff0353b2e8d74a35d1a933d0d64424d349830","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-25 19:05:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83152275463f624fa6ab0e5b8a00338cd6c99316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:34:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/010c4272d55542d59a5bf985d3dde61f : SUCCESS in 19m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/51ed0d924ef94109bd9511253268f3cd : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/91ecd4a5c9f146e8b6c9267c748021a7 : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/571b5c5dd23f47e7b2d3cb93801b2857 : SUCCESS in 1h 36m 28s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/45d6df1a8acb4664988586493ec4bb66 : SUCCESS in 1h 27m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4db199f868d84fcabbbba2c38d4308d3 : SUCCESS in 1h 47m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/367b985562474e9c955f83295f15187c : SUCCESS in 1h 53m 15s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d1bfa37aaab452288992f1ae1a4b6b9 : SUCCESS in 2h 06m 43s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36f1151b19b849528448eb7e5649dc26 : SUCCESS in 2h 17m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9805860199ef42d8862c33039e1d7b3b : FAILURE in 35m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8231dda6a9ca4ca2b435014b8cc9c5d6 : SUCCESS in 1h 01m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b72f16836146f78073fa2f6c858b7c : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a2f4ec7280e4adcb273aac81e8d1e4a : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"01f32daec33f1ba15ef1fc1eac118a9ee0d4113c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:37:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63b9cb2ef0f6969586d310e03184779c60fea9c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 22:47:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eecb8a8bc8fa4df6891ac82ccbb41083 : SUCCESS in 12m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/544d23628d354fc4841cdcd321bec34b : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/543fe04c494c4708a934936487b79480 : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/636eaf822f1e44d6a0e4d4375c122d01 : SUCCESS in 1h 26m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/991beb398e974eb08a221f372b1f3af0 : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bce6eda31c9845b58001f43657facbbf : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/427cbb7a01134013858f63031daa7525 : SUCCESS in 1h 57m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e37b4af52564bcaba19353f5706a7dc : SUCCESS in 53m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9639f65007405f8d82c2c71c538ec3 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13c38508cec34de0804640f8c7c5caac : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6b50f05ba1684ee70612234289dbf2e5f08a22af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 22:47:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"24cfbd7fd98e74970104f70de469cf285b09a495","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 22:50:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/519ec812ffb94f4392be701af347e148 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"887a59256eeba7a346b63d5c0db6d1f909d6b26c","revisions":{"887a59256eeba7a346b63d5c0db6d1f909d6b26c":{"kind":"REWORK","_number":1,"created":"2022-04-12 10:56:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/837550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/50/837550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/837550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/837550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/837550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/50/837550/1"}}},"commit":{"parents":[{"commit":"a23f92c9c728a3c0af53d9787cca927d43be3a6d","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a23f92c9c728a3c0af53d9787cca927d43be3a6d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-04-12 10:56:38.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-04-12 10:56:38.000000000","tz":120},"subject":"Use systemd_mount native syntax for mounts","message":"Use systemd_mount native syntax for mounts\n\nCurrently we have bunch of limitations related to the format\nof ``glance_nfs_client``. While systemd_mount role is flexible enough\nto allow mount cephfs or s3fs, variable format has weird assumptions\nthat we want to change for better flexability.\n\nSince keys of variable are changing, new name for it was picked to\nreflect purpose of the variable better.\n\nChange-Id: Ic0d91a3a873b4253255beac79becf01b4a304695\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/887a59256eeba7a346b63d5c0db6d1f909d6b26c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/887a59256eeba7a346b63d5c0db6d1f909d6b26c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
