)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~866686","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ieeeb9bc3b8922700e0c70f150cf1af0a75c805ea","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/distro","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-10 18:57:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieeeb9bc3b8922700e0c70f150cf1af0a75c805ea","subject":"Stop using need_service_restart local fact","status":"MERGED","created":"2022-12-06 11:12:11.000000000","updated":"2022-12-10 18:57:54.000000000","submitted":"2022-12-10 18:57:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866686-osa/distro","meta_rev_id":"625ba3cbcd3d3e1845934035acba23c46822b885","_number":866686,"virtual_id_number":866686,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-10 18:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-12-08 20:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-08 17:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-12-08 20:36:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 14:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-08 14:46:42.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-12-08 17:37:06.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-12-08 20:36:05.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"7be3fe3ed8c741ffa7c5562be58204be98b0cefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-06 11:12:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d838512d42c454a04a1a17cf9046040b6adc046b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-06 11:18:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"dabc06c483b4e7e2a3b887dea79a7eb7390377c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-06 11:56:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24953867108e98d84441fdf1cda15d5e333426d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:37:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33e3a78cbaf94f4a8e286ff42b42cd42\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6695d9b645747dca60d9635f9b624aa : SUCCESS in 17m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a2375c69ae4d4a2a8036333f3f10587f : SUCCESS in 1h 29m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4d5dc12e7604cc0b9942098e49c3e36 : FAILURE in 1h 26m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ac9d874bc2f48e3bcf1c79f04c52c91 : SUCCESS in 1h 27m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b7adbd4e0824154804433401944a74b : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9c483e1e96d4a41a6bfdedb9945b7c9 : SUCCESS in 1h 29m 35s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdc877275ae8415ebdc3dbe80f30c0a6 : SUCCESS in 1h 39m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1f39110c1f44052a2ad45df5f6a2329 : SUCCESS in 1h 59m 04s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35233aafdfaf4466b418056ab16a4342 : SUCCESS in 2h 31m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f55a865305fc486184cd900206f7ad4c : SUCCESS in 2h 05m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/158f7c16e11848169776f155a7f34637 : SUCCESS in 2h 15m 31s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d913274928b4cf0abe49dc38e7757c5 : SUCCESS in 2h 19m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/334f5b155d3f4064b79d6ea7e0570a11 : FAILURE in 53m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd2f713ef22540539a921755706745c0 : FAILURE in 44m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b68a21b21b34377a11ac71a50636ce2 : FAILURE in 37m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407e275b9d1f4f8a9f7d473ee8c86bac : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3},{"id":"56ca59f0c90210b6b1d5882a8e7046ede9ca2d8c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-08 11:23:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"43e82a2d4cb5b47884d526fef8077342c04032a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 14:40:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd430de9982946d3aaac61606475b2d9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08ac0412f5554e46b42dc9652dfd104d : SUCCESS in 11m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/16871fdfd1054f659cfe2350b28098b9 : SUCCESS in 56m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f366cdccbdc14e5a8ce7149040e69e18 : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2aac64d1ccd94821a4d5d7b4e72c5dfe : SUCCESS in 1h 34m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f56b0b3068354808880dde0489b076d9 : SUCCESS in 1h 30m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebb56ad40faa429697bdf136df66e69c : SUCCESS in 1h 15m 48s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/06f2f780e0f04a01a3343dcab0ad2aba : FAILURE in 2h 04m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/086a15f793534a2098622275b8931699 : SUCCESS in 3h 01m 40s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c2926495d044fc695d27be876e1eb6c : TIMED_OUT in 3h 07m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c910cf67056e448ca8b78aa617320408 : SUCCESS in 1h 57m 27s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b789c5a0c1ab4597a8ccbca989981d83 : SUCCESS in 2h 35m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0fb3f45493b4fac8f59c27595b08dd0 : SUCCESS in 2h 18m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/370a838c9d4648678a709e60ea721894 : FAILURE in 37m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/357c97c31c804759b6d227dec16acfe1 : FAILURE in 39m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d4f64a2ca6044de85410918c740924f : FAILURE in 37m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ebc438ba3bc45cd8f757764697138b5 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"62a3a6b4a48017142bdfeba83905b315b7f462d8","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-12-08 14:46:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4f59aae44408577ff2579e20ee33f9820c4cd93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 17:23:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53eefbe0b76d4f29892eba56f67a8a72\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d91bdfdb00b47b59cec8ccd0c0dbb0c : SUCCESS in 11m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e9177e1471e64f41ae7722b47d4e358c : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4cb1d4243d9466787bfef5ed2aed7e9 : SUCCESS in 1h 21m 21s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4723e600fc824914b6cfbf746680bb36 : SUCCESS in 1h 31m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82eca2771dab4ef9b970ea8bc64726e9 : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ecdb2d707dd41f0ac6b5c31dbccfa64 : SUCCESS in 1h 19m 00s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4201fd8ebed841ff9b94ca0b2688c7ac : SUCCESS in 1h 47m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cca1e20ec134474ae49eb92e7b643fa : SUCCESS in 1h 55m 10s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0d926e373ca4a38acfd8fe7ee67e1d7 : SUCCESS in 2h 14m 34s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42dc983499f94bf4993c85d34e7e9fff : SUCCESS in 1h 59m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12c2fc0a8a5743febf634910a6173a3a : SUCCESS in 2h 30m 07s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d2ca62d2efe4544b12a68d12806f0be : SUCCESS in 1h 54m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e7f4ecddccf4a52a1a258479e120753 : FAILURE in 34m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aaf6fcb3936d489290d6f11fdcc04253 : FAILURE in 39m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2f9db2e59be4e6087d2868e649ea0ba : FAILURE in 38m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b890cc7d3346afab11ef1c0f707c55 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"1fab91cce892745e323e2661e458ae95eb9e4e1e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-08 17:37:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9cd33702180e95b6b41b23e81c3c8b6c974bedb2","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-12-08 20:36:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b644a519c4171e3d722f36e241433b9bb029b35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 20:37:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"48e44cf1663a365c9bad36c78671a364bf71748f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 22:41:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64a1c6b436c446c4b0988a92b38e364b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f33ea5c82ca4b2c8147cea51887b93a : SUCCESS in 9m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c89055d737ee4b09b72d6126853e188d : SUCCESS in 1h 10m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/914e1984e6714b60a44f5d8fd6de0498 : SUCCESS in 1h 21m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb786c362d2e4f97b151eae6fda39336 : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c5b084ba4ac4c5682b3f1e57208085b : SUCCESS in 1h 20m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c809d5a00e8c4cf094c37912a14d4d57 : FAILURE in 1h 36m 39s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/214345b7e459411a9ab249d1bba56f15 : FAILURE in 2h 02m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e71242eac8a4133bef021922d70788d : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e9c40b4abb997247a707a0f07be46dd425235128","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-10 08:44:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ae65d5d55b96ee90941a92a67a3d633721eab24","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-10 08:46:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03cfc32ac5099d989085c45d411c27bffd330bfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 10:57:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3ea387d88f745969b2f33b35df714e3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61650bc609b345cc9ae602b5db7805cd : SUCCESS in 12m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d5b59de2dd4348d5b2294e9abb6fe71d : SUCCESS in 59m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c37349a18564caca2a730800cfacd1a : SUCCESS in 1h 14m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa202adedf644e97b1815f158464ab42 : SUCCESS in 1h 24m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd749c0e35c4676b9425aa49640edc0 : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/485ab5cf81974170b7ca3517d0481756 : SUCCESS in 1h 12m 19s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8537b729169248338390db1d7d9750c9 : SUCCESS in 1h 36m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f3d2233d5d4428da5eac4be00f00d01 : SUCCESS in 1h 43m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b6930b057fe4650883975e1a663259b : SUCCESS in 1h 56m 41s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6509b77d45b4fd6af1d614ca2be8101 : SUCCESS in 2h 05m 53s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd83363e967a45f6ba19da74b73b144d : SUCCESS in 2h 10m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fe83f24ed4f482cb0004a2ed4039259 : SUCCESS in 1h 37m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4db2ecb01c143c983af51bb35a2c802 : FAILURE in 40m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5af81449c92e467f83feb3ebfe7315be : FAILURE in 35m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706a5be6afbe4417a1d9d5271a051400 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f0d4e701e44f5c5509bd25df7e00d755cf8d888c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 10:59:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0d8ad7f0594506b044840b9207440c02e7b6102f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 12:56:31.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ca193762ac54b709ce43f8be110feff\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a51a363dd7664bcf9a795dcd592b9d7a : SUCCESS in 10m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0df9737cb00f489a8b4131c6e9dec943 : SUCCESS in 57m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4b6cc84821b436b9a1754446fb141ab : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6450a47a2b2641c8b07b9051816cf243 : SUCCESS in 1h 24m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e60688b6a3de4026a7b2ebfa02820bd6 : SUCCESS in 57m 42s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77ffdc4596384278a6efdadd53796a36 : SUCCESS in 1h 29m 32s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/092d3c422d874575a238f7cc2383474b : FAILURE in 1h 54m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c2fbf5b56246b19f954a15e9ca61ef : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b039c86a9f679d66e7cdb1a557e55c998a4aa874","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-10 14:18:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a66df43435632989c733573586a4407577fb818c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 16:47:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f955184eb24f90a29a6a05c1225cf0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec630f37190141fb807a7909c6c36b9e : SUCCESS in 12m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6e13708170af49adaa4bfd04b9992173 : SUCCESS in 1h 25m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/021ec48ac0ea4c70829dfcf574484972 : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/284ceace3fef46ff9f78beaed370dd57 : SUCCESS in 1h 33m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e06eeaa46b5d4476a0149ad30097e1cd : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45afddd200864f64b57cb5a3cc181d2d : SUCCESS in 1h 27m 48s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e557091bb07b4ca8b090a1bdd71f6ec1 : SUCCESS in 1h 55m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3456d5dbbe5d46a4841c56946d045650 : SUCCESS in 2h 12m 41s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49bf4df771404c4395e8a426c2700019 : SUCCESS in 2h 10m 40s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b57f47c8ade04468ad4994b32fc53bf6 : SUCCESS in 1h 26m 02s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/253e24f06dfd4b1699f34849efd0ef23 : SUCCESS in 2h 25m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98b68442b37241a4b7e83acdb8d4da91 : SUCCESS in 2h 01m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d43f4316e25a44ad90b2cceb051585aa : FAILURE in 39m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1814fe152ed04da79318e38514134c87 : FAILURE in 34m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3112db3fc534bcea78924b9bb071947 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"90ead0aebf35203a87f3bd75757b4c921cecc497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 16:49:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"10c8d4f974d8c2e8f666169406039cef8373a979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 18:57:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1c08ce8d064d1e991baf09761c7775\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf9816565d2343e7ae8d55e2a9633991 : SUCCESS in 11m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/667c54ba43cd44d0abb0f0f36bef06ef : SUCCESS in 1h 19m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/203d5a7f50b64d4fa259b0accaf42b17 : SUCCESS in 1h 24m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/505fdf862bbb40aa819f230efa3a4b42 : SUCCESS in 1h 33m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fb6af8e11244ef98bd4850aaf8d8899 : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3111a9ef897b456bbcfe88603811bc4e : SUCCESS in 1h 54m 20s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dcf9c17fb2d4f28bec4b16c66bd8563 : SUCCESS in 2h 06m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1dd443f6b9242ada8ebf91f00eb345a : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"1c4866e538f7748f841cdda62461d48ae4a291d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 18:57:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"625ba3cbcd3d3e1845934035acba23c46822b885","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 18:57:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89cc4a484a644d679e19c8237be61a9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472b3ed20e394562b96a166985dae45e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee521f91f281535b38047d94f32c987b9bd87ff5","revisions":{"8df74560f086b0b6c145e1c75c459ebfb6464d7f":{"kind":"REWORK","_number":1,"created":"2022-12-06 11:12:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/866686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/86/866686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/1"}}},"commit":{"parents":[{"commit":"504e641ed41c4447c0341f705a0aeb24c4069273","subject":"Merge \"Remove redundant vars line\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/504e641ed41c4447c0341f705a0aeb24c4069273"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-06 11:12:03.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-06 11:12:03.000000000","tz":60},"subject":"Define venv_tag as separate task for distro","message":"Define venv_tag as separate task for distro\n\nWe do define venv_tag locally using python_venv_build role so no need\nto do the same as a separate task for source installs. Though this task\nis still needed for distro path.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/862924\nChange-Id: Ieeeb9bc3b8922700e0c70f150cf1af0a75c805ea\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/866126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8df74560f086b0b6c145e1c75c459ebfb6464d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8df74560f086b0b6c145e1c75c459ebfb6464d7f"}]},"branch":"refs/heads/master"},"320a2c8ae62ee017728e3cd3b93e99a155cce6e0":{"kind":"REWORK","_number":2,"created":"2022-12-06 11:18:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/866686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/86/866686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/2"}}},"commit":{"parents":[{"commit":"504e641ed41c4447c0341f705a0aeb24c4069273","subject":"Merge \"Remove redundant vars line\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/504e641ed41c4447c0341f705a0aeb24c4069273"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-06 11:12:03.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-12-06 11:18:28.000000000","tz":0},"subject":"Define venv_tag as separate task for distro","message":"Define venv_tag as separate task for distro\n\nWe do define venv_tag locally using python_venv_build role so no need\nto do the same as a separate task for source installs. Though this task\nis still needed for distro path.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/862924\nChange-Id: Ieeeb9bc3b8922700e0c70f150cf1af0a75c805ea\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/866126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/320a2c8ae62ee017728e3cd3b93e99a155cce6e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/320a2c8ae62ee017728e3cd3b93e99a155cce6e0"}]},"branch":"refs/heads/master"},"ee521f91f281535b38047d94f32c987b9bd87ff5":{"kind":"REWORK","_number":3,"created":"2022-12-06 11:56:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/866686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/86/866686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/86/866686/3"}}},"commit":{"parents":[{"commit":"504e641ed41c4447c0341f705a0aeb24c4069273","subject":"Merge \"Remove redundant vars line\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/504e641ed41c4447c0341f705a0aeb24c4069273"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-06 11:12:03.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-06 11:53:09.000000000","tz":60},"subject":"Stop using need_service_restart local fact","message":"Stop using need_service_restart local fact\n\nWe can reduce complexity of logic on when to restart glance as we\nanyway leverage in-role handlers to restart service when needed.\n\nPatch adds also triggering service restart when install_packages\nare changed.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/866693\nChange-Id: Ieeeb9bc3b8922700e0c70f150cf1af0a75c805ea\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible/+/866126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ee521f91f281535b38047d94f32c987b9bd87ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ee521f91f281535b38047d94f32c987b9bd87ff5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
