)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~873632","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I87a5a1bb5028706e1768f53f53c4567cbdff64c9","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/quorum_queues","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-09-04 08:19:43.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-02-27 13:20:31.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I87a5a1bb5028706e1768f53f53c4567cbdff64c9","subject":"Add quorum support for glance","status":"MERGED","created":"2023-02-13 17:37:40.000000000","updated":"2023-09-04 08:20:46.000000000","submitted":"2023-09-04 08:19:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873632-osa/quorum_queues","meta_rev_id":"1bfbdd6df55d3f9255d547e4c3f3f511d1488fc5","_number":873632,"virtual_id_number":873632,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 08:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-08-22 21:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-04 06:16:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-04 06:16:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 19:24:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-14 11:02:19.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-08-22 21:26:10.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-09-04 06:16:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"93fabe025967cc56862a9d7a8b1cb000252c4a95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-13 17:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5233d45abb0410fead77fc5a7f70d751eb801d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 19:24:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/680521f6877643b5a7e29570dd4b477c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7e9968bc36046419a1fb7cfecc38da6 : SUCCESS in 17m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8c10a4893c984c159a2eb5a543dd7826 : FAILURE in 37m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a534904bab14010991adaf18cc74b49 : FAILURE in 37m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ef55aa07b5c4197960c8e0c75f93198 : FAILURE in 42m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4ccd83b996343fabd544dee519e35bd : FAILURE in 39m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a1aa0d065e84161889cb24e7a5cc7c7 : FAILURE in 52m 30s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8679fb69f67e4d4694e53ccb7f759e14 : FAILURE in 1h 27m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aff68bd14d76448cb20f920a9a7a2142 : FAILURE in 1h 40m 12s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/894b1286d2584063a2425f095942c526 : FAILURE in 53m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/084cf7fe09c44971935d14ef90b8db2f : FAILURE in 50m 19s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f8fc93c983f41559339bf2b9bfd2825 : FAILURE in 1h 04m 01s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/440f4fb62de540ff886b920ec8573856 : FAILURE in 1h 05m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/eaf67c9073074ecca956006f176d4c55 : FAILURE in 35m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e48b5d89eb674a53a6f7289de45f85f2 : FAILURE in 35m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4b9ec902a8452a873eaacc67a974be : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f71968d8eb1f68f4ecb883bea17435700d4eb556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-13 20:35:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"edf161c5fbc8c6f6508a2189221bfae259949197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:47:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a28d4e2ab7b3413f9dc3b8aea6b7d4e4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cc2831bd6dd4e82a87b77fa38895517 : SUCCESS in 12m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3b77b9f775dd4d04bec3abe754a16d6e : SUCCESS in 1h 09m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0263975023a3439e9e0eaa47466083b8 : SUCCESS in 1h 12m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3468c823eaed4732b0990fd15efa8ed0 : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ca8017742b1498d92465d7ca1f04979 : SUCCESS in 1h 51m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c91bd3cb66944829fc26ea40f68e9ab : SUCCESS in 1h 14m 25s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/acb780136ac34fb787fe1227c8b6b1db : SUCCESS in 2h 13m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba0eab202ff542459fc609b00d85c620 : SUCCESS in 1h 47m 00s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/231680e1267342d585adcb05dd923545 : SUCCESS in 2h 59m 32s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d9b10efdccf4553a6f1299cccd68c09 : SUCCESS in 1h 35m 20s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e4465d15b6d44579f47e19926f160dd : TIMED_OUT in 3h 08m 13s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc50030a8a5f4d6484c32c426a19c8f9 : SUCCESS in 2h 21m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b552d0be25114f0b9fd15224ea00d9dc : FAILURE in 38m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b70e2380aa71478b887af6ff1e187659 : SUCCESS in 1h 26m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47bdbc8ec5af40ca8c65b825e725bdc7 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4f9bd2bcb842a9c0a9b7965bbcfcda67308766fd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-14 10:10:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"974d8a4510e9253365e7d10a405af802ae439a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 13:23:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1625efb5b947fe868d644b2362a365\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a0a5b878d843e88e878e7048a0c96f : SUCCESS in 12m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/aeccc5a20ed84f5a8ca8fbdaf3ae0c17 : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dded9a2993df4edfaa149d97ed65990a : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ee3417d173e43edb614dce9dc169f3d : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27f9d806095d41a8b35fde8dbc3b4d13 : SUCCESS in 1h 44m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/403889d05eb647a3a7bde22b199d06ae : SUCCESS in 1h 20m 19s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcbdfdf84cdb4545836543ea0cc5383d : SUCCESS in 2h 11m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a6b55bd259049b0b31d58b515898c3d : SUCCESS in 2h 03m 23s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/375926a5470a41eda666f336a9c56308 : SUCCESS in 2h 12m 50s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76b2eb5e0ff9494cbc73a8ebac9b2d1f : SUCCESS in 1h 47m 29s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8c6b3593c36447e85bc58c0c2b2c494 : SUCCESS in 2h 45m 58s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8ddda1c59aa402bae3b1e70ff31f83c : SUCCESS in 2h 32m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c33aa4fa8744350ba06e010531b1494 : FAILURE in 37m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9a37f7de8584e20ae84c5cd91303f72 : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eafeb7d8fd44b53aaac3f8535afc08f : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e70fc05bba8403aa3ef2f91d1d219171c70c903c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-15 17:02:13.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"741dad9289f09298a75787eb2cc2b611805c6914","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-27 11:28:23.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"106fb0a8abe8641203503244e459532d26a5bdac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-27 11:30:15.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"34eceda20f87300e6d7f7688fa7383ec709ea1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 13:20:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b70dafdd8712454ab314b0b1cc7b3a79\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a5a0c9925b543e18661de919ab3e89c : SUCCESS in 11m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bdbfe95c071148b6849f506639920879 : FAILURE in 46m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/75fd8ae0b0534a85af5b40f2fe963af9 : FAILURE in 49m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e225126c6db040b9a54c9d3f3aa673b0 : FAILURE in 42m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/983a215681ef4288bfdb2be33be4bb83 : FAILURE in 40m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e900e3c446374622850085c16395be5b : FAILURE in 46m 04s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/805cadb5a80a4d3fbefaa168628766ef : FAILURE in 1h 30m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3412c55feb449ada20af9ecbdc87bf0 : FAILURE in 10m 00s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95d291b1d0264949b4d74ebe01067d32 : FAILURE in 50m 05s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3054816e6e3b466eb39f909a4f5c0e5e : FAILURE in 57m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e76f09496654b32a9afd70414bd1669 : FAILURE in 1h 02m 44s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f868a513bb742819b80f85c7a63776b : FAILURE in 57m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e14dd51531b14970974e7dde232ec262 : FAILURE in 34m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6fcd6e99ead4c0493d12a91b803a86d : FAILURE in 33m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa4523041a24bd38e43d37d1b24ef95 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d3f577db498cf06d038d677cfabc373761b3f9f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-31 17:00:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cf5f2737cdc47b0ee63660816d6c527a2dc7c707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 18:45:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77f6d7611c544dddb5c7c713905a261f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb1ea3681d384d7a89a5019359397a1e : SUCCESS in 12m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/529df8ab18574cac9f3cdd7b6f25c1a0 : FAILURE in 44m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ea3ad8bef004a0090a9c2305183b3ef : FAILURE in 49m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd09ccf432c44819b1f1dc40d57db900 : FAILURE in 46m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53f53ed5e60b4d7c81b403f08271633d : FAILURE in 46m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/574f31dfb748469888bc61f4ef1dd370 : FAILURE in 42m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/658dc9a07c1848e9bbed4567bcb4b1b9 : FAILURE in 55m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c01cae7f1b224251b8a23aaef952f93c : FAILURE in 43m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75963c828a6940309c7804a7ac9902ee : FAILURE in 1h 39m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/028c744ddcfd4f6294b812edb8d40204 : FAILURE in 1h 17m 51s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea99800d22bf40288321ca81d1a9b291 : FAILURE in 1h 22m 19s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4b2ca766f694e7e9750932bdd5501fa : FAILURE in 1h 06m 54s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7370fcc22ad44b8b217f433e31333d8 : FAILURE in 56m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/046495be731a4e84bc01805d85502113 : FAILURE in 55m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00aa72e5cbb341ae94825aa2d4f76626 : FAILURE in 1h 16m 23s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6159fd144b364b9a94861952f8926128 : FAILURE in 36m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/feae595a1de64fbeae18dd0750d0d462 : FAILURE in 38m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2bd84e2c42644d99e82c42fe956c616 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3b22ecfe3ef1e0e58e29961ba02aeec14fdb4bd8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-01 11:55:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32e736a187236fe7f6f2a751fef30ec0656d508d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 17:01:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f25fc9a909d4422a36f917060ef9ed3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b7cb04541e149c9b65ef58f80085457 : SUCCESS in 17m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/881f0cdbcfaf4cb5809462689076b5be : SUCCESS in 1h 40m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/82edb5d66ea3434d83a24c660626fc90 : FAILURE in 1h 03m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4cf209056074bf4896bc6cf810f1179 : SUCCESS in 1h 35m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/401810ef290b4b75ae51326336d498ca : SUCCESS in 1h 35m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a34da8170b04f4f901aa18cbba31546 : SUCCESS in 1h 37m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad4aedb5dcd9423cb74f15865d27eb87 : SUCCESS in 1h 44m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef108b4b7ca84d1e897a0b127ec5a6f1 : SUCCESS in 1h 36m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11bdc499d2f045eeac10f3cc03058d40 : SUCCESS in 2h 11m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58925ab210d448ce9dd4302b220450cf : SUCCESS in 2h 13m 48s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7af7548edda24c95965ceaa31e787e6f : SUCCESS in 2h 10m 01s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7ee68d43cf243008c23cf5d3b1a1107 : SUCCESS in 2h 32m 19s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad389bb558b9422ea53f79d1df0df1ca : SUCCESS in 2h 14m 09s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0e50e0fd42044ffa9b46a73bcd69ba6 : SUCCESS in 2h 44m 00s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3690dc6837e43e6acc621e17e1ca14a : SUCCESS in 2h 10m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1414ae91809548e7acac077964fec5e5 : SUCCESS in 1h 08m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6e3701c86f64d15b79fe2f4f6e3a69d : SUCCESS in 1h 09m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d02dead37946b79110fd91e426c73a : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"5e6a50efa4d8827c51f4e506a2b43a65da95d0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-06 12:05:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c9a00624e1c405821b87f920b26ba2884cae9cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 16:33:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f25acee6afda469e86508eda571e0727\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd53bb76c6a548da99ff48c1de2dcfd9 : SUCCESS in 10m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3b4f39912df44065b5046faee74a2f7d : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5d7f3fcd1dd4a9db5abcd68ee44874f : SUCCESS in 1h 25m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f065d35cd5df464da75a3f7a054f68e7 : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f2232fce4a6457d8e7ce9c8b6524f49 : SUCCESS in 1h 41m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd49aeb3af9b475c87da3929267de223 : SUCCESS in 1h 42m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/436695860aca44e48efe918180212be5 : SUCCESS in 1h 50m 10s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3080fe86229641ecb948441fc3cc7e06 : SUCCESS in 1h 48m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f26b7465ca6410e9af72049ea0ea8f5 : SUCCESS in 2h 38m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a4ec66f504e4016807896c824733ef3 : SUCCESS in 2h 08m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1afd839b450244acb71bb01d075a71e0 : SUCCESS in 2h 13m 22s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb06a9217863479ab083688c7df4ba7e : SUCCESS in 2h 25m 29s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8f45571aa55472280215ad43d6bab33 : SUCCESS in 2h 37m 47s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae2ff09d9f13485195f6cfc3385f7a88 : SUCCESS in 2h 13m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9442a361873443548a247abe0068caaf : SUCCESS in 55m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba46f00336794c86831c5890a0165739 : SUCCESS in 55m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da83a0ae006b4ad7a6ea2ace4bce1c12 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":4},{"id":"b71dd164120d4366590f3eb617679b82bc218f94","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-22 21:26:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a2b009487f8e49fc2d08304bc0ddd9993dccd1fe","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-04 06:16:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fa18f5308c6a57572bddb033b61be2a247b9abfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 06:18:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f6249cb942da7fe54b5e48112f3838f6dc95bff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:19:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5686bb462e14e79a3a9c3df7fa4d2d9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/997a8054b32347b4bd941bd7520140c5 : SUCCESS in 12m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/59de8fcca5e944ea879b6403a72799b7 : SUCCESS in 1h 39m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4418844068c94ace8d02f8555a3694e3 : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c292bc00a7734a4895fb92eff9ad9ca1 : SUCCESS in 1h 23m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ebae506dbece40d5889b6b741ad7973f : SUCCESS in 1h 39m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dbb12dd07564e5d9e976edd34ed1d52 : SUCCESS in 1h 46m 54s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e9c68e9b1c14656ae0e47624572bff5 : SUCCESS in 1h 55m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fac24cc10b94b319e4d190961a5dce8 : SUCCESS in 2h 00m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140a531f6fac4a6f8681a77140b38ee4 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":4},{"id":"90b5e09c921ed7f50dec431ee1948cd58a1ca634","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:19:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1bfbdd6df55d3f9255d547e4c3f3f511d1488fc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:20:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7ec86ba22b4aa69c7a88b96327855d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8988c332074a6eb88690fe63c8d325 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a5d922e4ed9e83ef7cce99d17f2b7ea6f47c1b05","revisions":{"96dfa4653c04f1c5b95f1d129b954b2bc78c20ed":{"kind":"REWORK","_number":1,"created":"2023-02-13 17:37:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/873632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/32/873632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/1"}}},"commit":{"parents":[{"commit":"7b88f5a74e1f94d2ae30c391fa87de221704b930","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7b88f5a74e1f94d2ae30c391fa87de221704b930"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:37:02.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:37:09.000000000","tz":60},"subject":"Add quorum support for glance","message":"Add quorum support for glance\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.messaging/+/873630\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/873618\nChange-Id: I87a5a1bb5028706e1768f53f53c4567cbdff64c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96dfa4653c04f1c5b95f1d129b954b2bc78c20ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96dfa4653c04f1c5b95f1d129b954b2bc78c20ed"}]},"branch":"refs/heads/master"},"aa4f43dbceec7c240e4f9c45a160bd9bdd7e8b8f":{"kind":"REWORK","_number":2,"created":"2023-02-13 20:35:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/32/873632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/32/873632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/2"}}},"commit":{"parents":[{"commit":"7b88f5a74e1f94d2ae30c391fa87de221704b930","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7b88f5a74e1f94d2ae30c391fa87de221704b930"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:37:02.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-02-13 20:35:10.000000000","tz":0},"subject":"Add quorum support for glance","message":"Add quorum support for glance\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.messaging/+/873630\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/873618\nChange-Id: I87a5a1bb5028706e1768f53f53c4567cbdff64c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/aa4f43dbceec7c240e4f9c45a160bd9bdd7e8b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/aa4f43dbceec7c240e4f9c45a160bd9bdd7e8b8f"}]},"branch":"refs/heads/master"},"313e743d821e75a2fb324fbacab0b7bc194704e0":{"kind":"REWORK","_number":3,"created":"2023-02-27 11:30:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/873632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/32/873632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/3"}}},"commit":{"parents":[{"commit":"7b88f5a74e1f94d2ae30c391fa87de221704b930","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7b88f5a74e1f94d2ae30c391fa87de221704b930"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:37:02.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-02-27 11:30:15.000000000","tz":0},"subject":"Add quorum support for glance","message":"Add quorum support for glance\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.messaging/+/873630\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/873618\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/875399\nChange-Id: I87a5a1bb5028706e1768f53f53c4567cbdff64c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/313e743d821e75a2fb324fbacab0b7bc194704e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/313e743d821e75a2fb324fbacab0b7bc194704e0"}]},"branch":"refs/heads/master"},"a5d922e4ed9e83ef7cce99d17f2b7ea6f47c1b05":{"kind":"REWORK","_number":4,"created":"2023-07-06 12:05:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/873632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/32/873632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/32/873632/4"}}},"commit":{"parents":[{"commit":"1f989219b404934164ec0107bd0e10b00ad6628c","subject":"Merge \"Add TLS support to glance backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/1f989219b404934164ec0107bd0e10b00ad6628c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:37:02.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-06 12:04:59.000000000","tz":120},"subject":"Add quorum support for glance","message":"Add quorum support for glance\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.messaging/+/873630\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/873618\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/875399\nChange-Id: I87a5a1bb5028706e1768f53f53c4567cbdff64c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a5d922e4ed9e83ef7cce99d17f2b7ea6f47c1b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a5d922e4ed9e83ef7cce99d17f2b7ea6f47c1b05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
