)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~876328","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I5a52c0de14ee3a6215edb64dbc3bd48512d57e2e","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/systemd_restart_on_unit_change","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-11 20:37:35.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-04-10 17:59:53.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I5a52c0de14ee3a6215edb64dbc3bd48512d57e2e","subject":"Ensure service is restarted on unit file changes","status":"MERGED","created":"2023-03-03 12:59:17.000000000","updated":"2023-04-11 20:38:28.000000000","submitted":"2023-04-11 20:37:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876328-osa/systemd_restart_on_unit_change","meta_rev_id":"0f543ef7c67b384a0ae252064d36819c0e0bd8a1","_number":876328,"virtual_id_number":876328,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 20:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-04-11 18:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-10 17:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-04-11 18:11:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-03 15:25:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-04 22:50:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-04-04 14:02:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-04-11 18:11:31.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"3e403d7992a88de337b4ab8d5e753fa170ffedc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-03 12:59:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"714da0bfa75c9ee5431ca02e7c48de685401f35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 15:25:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23c282e50ae148cdb2802f357893fc0c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/214f8c884b934b02a726c207b88227b7 : SUCCESS in 11m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bde64eed9f5d4ac7997e76d081911067 : FAILURE in 42m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d1a824dac2042b9b91dc46a1bdb95e3 : FAILURE in 38m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f4d887bf49624617a9034781ff931f93 : SUCCESS in 1h 38m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/766b3a2cc172484ba270b84a578dbacc : FAILURE in 46m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a19de787d0a42738fb24e466ba7f830 : FAILURE in 41m 50s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a89b344a3dd49d08eefc9e06a853fc2 : FAILURE in 29m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4747bb4aa1b47699ee4a27dd9602d95 : SUCCESS in 2h 11m 47s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f59cf51ad00f4811872013c6d14f9aaa : SUCCESS in 2h 06m 51s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eda9d7f8e1334b3999b7b7293223dd0a : SUCCESS in 2h 08m 05s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/816a58d5579f4568b2239a1d4f2e80f2 : SUCCESS in 2h 23m 58s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffdee382329c467a98d4a77f1b6e54c2 : SUCCESS in 2h 18m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/af6b73a4999047cf8e538ff316caadc0 : SUCCESS in 1h 12m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12dc1c14889949488dc1dbae8b01bf0e : SUCCESS in 52m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a155b0582b6045a591a519ad09131063 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2dfde96fea3a3998d5a5e1b5ea332627f48dea8f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-03-04 22:50:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14febb153160c286c8878d597bb657d1b0d46d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 00:51:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/857fcbdd1f3d43e2b1815e613643ec86\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c14dbd35cb934a4e980548ef7479f264 : SUCCESS in 11m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/891c9869fe564629b2920643cd49e7f2 : SUCCESS in 1h 32m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e06cf8edae549ea91a09c106d3a7de3 : FAILURE in 25m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ebbddbba7150432998b1ac4b653a9416 : FAILURE in 26m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1453f419206649d6aacdadd982d9f022 : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e913ef3962a49e696f103313926d487 : SUCCESS in 1h 19m 32s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fa36913fa8347779d7c2bea5c1390b3 : FAILURE in 28m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa842a5e91564626a645f5b387ab9894 : SUCCESS in 1h 50m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3991f7bf811d4408b673c797a17e3691 : FAILURE in 39m 12s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bb369defc444b7ebf54ec4c5b134792 : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be390700ee524b03aebf3bd7b549ae23 : SUCCESS in 1h 57m 18s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d2ff616202242eba3e29f1ca0ba5add : SUCCESS in 1h 49m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/63dfe6197ee04d1c9fd5fc91b9682578 : FAILURE in 23m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d49b9e87f1848fa8475ca0226d798e3 : SUCCESS in 56m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c1650788e54fa3bdb8acedda6ce18e : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"807a8ce6a543ea185483030ce4a0e3d295f89c59","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-04 10:08:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e61b4834ffa7c038a541af11877a07a0bbf2755d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 12:50:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1fa4e0e43043faaa6fec979373fb06\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c792a9fa17ff437dab7b2396bde3b4b0 : SUCCESS in 16m 51s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/20d3d58d27414baf87a9e6dec3210bc2 : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf00f472af104d3881817d1a16ffaf35 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/085a0cd412ea405b87ad3124f0b3ae6c : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a3a0f0e12024af7ba87030879503fe5 : SUCCESS in 1h 25m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6cf5b118dff474e9437f86ca60416aa : SUCCESS in 1h 19m 33s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/820c3e02e4a54de08bb6625e3c70997a : SUCCESS in 1h 52m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/567d3dc1236b4a08a7471e7f08bad7a5 : SUCCESS in 1h 54m 15s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/374b02ff309245e1a5660617657b3161 : SUCCESS in 2h 00m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3b03f55628248d2a5e91974106f9db9 : SUCCESS in 1h 55m 06s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22fe38d402d74700bd075b18a1b666f8 : SUCCESS in 2h 32m 06s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df8b7a077af24478aa766d7cdd0299fc : SUCCESS in 2h 05m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/da0eaaf4c6de4f2799ce2640f4400d91 : FAILURE in 29m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d85a5e960d846b997832c17e89799d3 : FAILURE in 22m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4090c9038248ef94a2afd1a009c35f : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d2ac6f69664c0114d0df327cde27a8223bc8e9e0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-04 14:02:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2c5b1ff375ce160683c9e2fe4ac2b6dfa4b63e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-07 08:27:39.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4264fea58aecc788c174c474eb6ed5c8c25e3ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 10:42:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/764151a3d8a44d23a6f96a941b56ddd3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52615a1c699e42d3af414fccf8c9c6e6 : SUCCESS in 16m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/61e97c8d4b1143f6bb0b4f3c01b71bce : SUCCESS in 1h 30m 27s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/33aef518ff314cad81c14346eb70c871 : FAILURE in 28m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2e0b58471ae43079b05961c6379e6e2 : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13c0e6d26a76493ea179281f5b625120 : SUCCESS in 1h 35m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/456b3c5024424767baa6d13d6edc6aab : SUCCESS in 1h 25m 44s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e7fc0041c18482db0d04ddf116b1de6 : FAILURE in 21m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c278dd400533406385ef3f47a1ec6c90 : SUCCESS in 1h 55m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ea3fcbb85cd4064be75d94b67a383fe : FAILURE in 1h 57m 36s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb75895decf047f6b8061a1494bdbf96 : SUCCESS in 2h 03m 11s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5227462aba1d49f4901c3f2a1a55d232 : SUCCESS in 2h 11m 33s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ee8bc65807d47aeabd37828725f8433 : SUCCESS in 2h 05m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e082ab8e3a184d26bd8d2e76eda62036 : SUCCESS in 1h 15m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4631ff8681c5432daaa2f36d99c00592 : SUCCESS in 1h 06m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f898172fcb46119b46c521fda910e4 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"456dbd9e29d59860b2a6f7ece88158f2a494be63","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 08:40:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c80a50d9f3ea1c97686be8a0781ef740b2c2c716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 11:24:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0a763afede94a2ba94cb9a3eed150c7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3db69e5e249c4ec7850e86a505dbf7ae : SUCCESS in 11m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b5afae9f1f3e40bc9020d1efd2581fb3 : SUCCESS in 1h 20m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e916464c32cf4a57b1713b78532baab0 : SUCCESS in 1h 26m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a8974d1d64a40f899ba02a643a182c1 : SUCCESS in 1h 38m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f37375fc7f564cbf85b0a0be2bd5f914 : SUCCESS in 1h 26m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e54f7fdce4541389af6d041ce32ec98 : SUCCESS in 1h 28m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28a6f0bebf1f4d408c2dc2ce8a1c4674 : SUCCESS in 2h 05m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43602428527d4e35b1d1da506285401e : SUCCESS in 2h 08m 15s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/875b69fbe13d4e2eb33ee311b14a5481 : SUCCESS in 2h 00m 32s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f07548052350420f8684328aa2fb9b25 : SUCCESS in 1h 58m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a41e63ad5f444a17a33336e02bb58490 : SUCCESS in 2h 16m 28s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b13e1a9dda246e185edfbc2cc20429c : SUCCESS in 2h 02m 28s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cca7fb2d209947a49e6754f4be7e5d37 : SUCCESS in 1h 03m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dfebb5750d24b65b93161ca80020b3b : SUCCESS in 58m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29053d36a8514111a746c78a7fa6294e : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"c39dec2720774f61721b04553fed5f03c6aa204d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-10 17:59:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"078f71c2b57a5cf2cb05408bf416aa4291cedb8c","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-04-11 18:11:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49f2c55dbbd58fd79803835d1cb5c630a53df16d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 18:13:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6723e4a9da5ead1214d5ad10346234b9987d11d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 20:37:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f101a61dfa754da39e6810ffbfd787cf\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe7459a7ccd349da876f2902e8a70a57 : SUCCESS in 11m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8725ee8a0c474a6a84d20f0c4293ac1b : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17534b6671184c18bf79b565c59172ab : SUCCESS in 1h 22m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a166d09de844bc0b5328a7912bed03a : SUCCESS in 1h 29m 26s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c945d13ca804b9ebe18f2a2dfb32c77 : SUCCESS in 1h 59m 27s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/203cf157e06944c3be541d05509baccb : SUCCESS in 2h 21m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f92e62d1524f2d88813e9079832bce : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cb0358839db5df9b49721dde36d7cd2fa41c69c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 20:37:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0f543ef7c67b384a0ae252064d36819c0e0bd8a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 20:38:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/706f8dc19a434f329f97b628673e7da1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001cac3c265e432c81e7fe4360bbfbd1 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7c7e4864e899cd2404cea3b08495221e37f523d","revisions":{"a954badc9549d2db7ee5e5226254ec9eda09bbea":{"kind":"REWORK","_number":1,"created":"2023-03-03 12:59:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/876328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/28/876328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/1"}}},"commit":{"parents":[{"commit":"7b88f5a74e1f94d2ae30c391fa87de221704b930","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7b88f5a74e1f94d2ae30c391fa87de221704b930"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-03-03 12:59:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-03-03 12:59:06.000000000","tz":60},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: I5a52c0de14ee3a6215edb64dbc3bd48512d57e2e\nCloses-Bug: #2009029\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a954badc9549d2db7ee5e5226254ec9eda09bbea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a954badc9549d2db7ee5e5226254ec9eda09bbea"}]},"branch":"refs/heads/master"},"f7c7e4864e899cd2404cea3b08495221e37f523d":{"kind":"REWORK","_number":2,"created":"2023-04-07 08:27:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/876328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/28/876328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/876328/2"}}},"commit":{"parents":[{"commit":"7b88f5a74e1f94d2ae30c391fa87de221704b930","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7b88f5a74e1f94d2ae30c391fa87de221704b930"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-03-03 12:59:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-07 08:27:38.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: I5a52c0de14ee3a6215edb64dbc3bd48512d57e2e\nCloses-Bug: #2009029\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f7c7e4864e899cd2404cea3b08495221e37f523d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f7c7e4864e899cd2404cea3b08495221e37f523d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
