)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~879698","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fyoga~I48feac2ea789782e55bd49196e631cd4df9778ce","project":"openstack/openstack-ansible-os_glance","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-04-11 15:32:10.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I48feac2ea789782e55bd49196e631cd4df9778ce","subject":"Disable uWSGI if ceph is used as a store","status":"MERGED","created":"2023-04-06 08:57:47.000000000","updated":"2023-04-11 17:56:54.000000000","submitted":"2023-04-11 17:55:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879698","cherry_pick_of_change":879370,"cherry_pick_of_patch_set":1,"meta_rev_id":"e478073700203fccf157208a78fef9c79f75c4e5","_number":879698,"virtual_id_number":879698,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 17:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-04-06 09:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-11 15:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-11 15:32:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-06 08:57:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-04-06 09:24:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-04-06 11:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e3e3fe8820ef65915ff97550d4c5ae00522c2f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-06 08:57:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"93c67bc3716816ab126f099c1a04bd52e246bf69","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-06 09:24:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b98cf106bc28cb94f0446ae8bb8dc569f90fe4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 11:20:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/225ab75f88d0498aa79bf79c7cb35fa2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac0efbd7b4ad4b729fd3e0f7d0d15ad9 : SUCCESS in 10m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ca4d03e00414081b28092acea1beae5 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/39132703c3ed494b888a2710ec78d67e : FAILURE in 30m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b278022dc2e44ec59815267e74a219e3 : FAILURE in 24m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d58b024f5614ec1a1ee9d74ede6dcb0 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/167426e202cb405dada000a47c470221 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51a01e0d6a6423b873706bd768342bf : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cf1603033bbf43179af6dad303405aad : SUCCESS in 1h 26m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/29285073782a44b48ec26e94517b9efd : FAILURE in 23m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c49ec93b544f491aa067729c5bf84288 : SUCCESS in 1h 36m 26s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8829a981bbc458b9e73ffb35ea36922 : SUCCESS in 2h 00m 35s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11ed70cec2ce4d73ba5a9155722888d7 : SUCCESS in 1h 33m 55s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d634b0fe56045ff85471fb35ef5610f : FAILURE in 2h 15m 59s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/752e322b75bd48989feb90cf7057fe68 : FAILURE in 43m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f64737d967604e1abd859f6b907ce2f9 : FAILURE in 31m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3370aad6f23843b48979070b3e9674b5 : FAILURE in 36m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/abfb740c2ef0491f834395d42f88be71 : FAILURE in 37m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4d74f330b7541eba19ac3f17d33da76 : SUCCESS in 49m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2ddbb797b649b78d22c2e0cff9ab14 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d53b1d6b854b1b366829210c60343f80fa180928","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-07 15:23:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a7d1305f5fe117780c8fa7fadd4b900755e2e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 17:46:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ed7743229c489d8d59ea73e3a21f87\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/222cf7609b5441049d049f14f026c38e : SUCCESS in 11m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/29a6f8f4fbb94a90bdc981a4ad8fa3c3 : SUCCESS in 1h 13m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/07714db690da4e0fbe3ce22947edc06c : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d319c6d338644f7e9758a0783f495125 : SUCCESS in 1h 16m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/582729e3c4cd472fbf5a5af7223b112b : SUCCESS in 1h 27m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/470493beac0942e780e90a748d4991ce : SUCCESS in 1h 21m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/facef4dc06844a05b94e7bbde45f2ba0 : SUCCESS in 1h 14m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4bc0e389b3894c62bdc883efd3bfcd44 : SUCCESS in 1h 24m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1b235c009924ccdb5e5ca03d2b81d35 : FAILURE in 25m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3927b662f57a4db889cdd6bb2565a102 : SUCCESS in 1h 57m 26s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d378c1384f14e6d952e17e24a4dcc43 : SUCCESS in 2h 03m 11s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e4d634997f640b1bff8e4be9ca62b44 : SUCCESS in 1h 50m 50s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9dd466fb7ee403a9919ce21a5797a50 : SUCCESS in 2h 17m 00s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cfb819c0c2644f8a49cc4dfff5c8546 : FAILURE in 47m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/24f84aee10b445f8969c38a8fbef1906 : FAILURE in 1h 00m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c4ee3e191c74cf09e97c892803d8e00 : FAILURE in 39m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63c1924f77054ff28942102500f37600 : FAILURE in 43m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e54872512704c349f04a325b5d04afe : SUCCESS in 54m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a205eeb44c184133a0021c61f7d22853 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c7c7a1f490a035c4831e7e3406546fb42fdf56ca","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-11 15:32:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"debfa020caa4ab27280747943aebf92cc274074f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 15:34:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99b000dacf9b8e9ed237a08d3823502d6b872c3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 17:55:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b9121ebec14bacb02787f0082135fb\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d69d1a0366dc45e2bfdd3fd19d8f489e : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/523eb3db059645d1a74317e59056fa42 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/94f04cda310544d7a9612995818c878f : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/903903dad22c487a8746e40299233271 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf616f34b8cd441d9898deb92a97e8a8 : SUCCESS in 1h 18m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/167977df7db446d8a977a2c158fab9f1 : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af35ccd26ff448caa20c687eec182d65 : SUCCESS in 2h 14m 42s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39698f08a84f4856be8afc94931f8436 : SUCCESS in 2h 12m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11774beeec884e77be806a304d0d1441 : SUCCESS in 51m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e67889369ea14519b35ab6c0af51dda6 : FAILURE in 39m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7a1ca9a92548f583afd8e7c0986831 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e302231c2a009bb69944f6f26d2eaa3ca5e28a98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 17:55:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e478073700203fccf157208a78fef9c79f75c4e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 17:56:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6819a3013cce42b29d31363146cbda3d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d9474c6a89409686de90df11f3b9b6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c5c5ca1622197e106e3885bb253c135aab8de97","revisions":{"2c5c5ca1622197e106e3885bb253c135aab8de97":{"kind":"REWORK","_number":1,"created":"2023-04-06 08:57:47.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/98/879698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/98/879698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/879698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/879698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/879698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/98/879698/1"}}},"commit":{"parents":[{"commit":"9438ed438a2fc4c94b376694555f8db4532bdabb","subject":"Update .gitreview for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/9438ed438a2fc4c94b376694555f8db4532bdabb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-03 17:48:46.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-04-06 08:57:47.000000000","tz":0},"subject":"Disable uWSGI if ceph is used as a store","message":"Disable uWSGI if ceph is used as a store\n\nAt the moment there\u0027s an issue with chunking in case uwsgi is used\nwith ceph backend.\n\nChange-Id: I48feac2ea789782e55bd49196e631cd4df9778ce\nRelated-Bug: #1916482\n(cherry picked from commit 295533132b6d143fe9d0499545fa0d396a633b49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2c5c5ca1622197e106e3885bb253c135aab8de97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2c5c5ca1622197e106e3885bb253c135aab8de97"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
