)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~892489","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I755822089bf406f532a718db5b84df553cbc2084","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-09 19:54:29.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-11-09 15:26:18.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I755822089bf406f532a718db5b84df553cbc2084","subject":"Use correct default for glance_mount_points","status":"MERGED","created":"2023-08-23 10:16:36.000000000","updated":"2023-11-09 19:56:11.000000000","submitted":"2023-11-09 19:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892489","meta_rev_id":"d247601e6eb92c6011147cda9823eb41be76225e","_number":892489,"virtual_id_number":892489,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 19:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-11-06 15:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-09 15:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-09 15:26:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2023-11-06 08:33:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 12:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 15:44:53.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2023-11-09 15:26:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"bb0e27d09c12d362ef4f13ed5298728315ec1a34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-23 10:16:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56917ac19474968dfefe75db56b141a129b7efc5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-23 10:21:18.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30084b97eb4d06c5bd4fd2713b0c0f6d8c80c5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 12:47:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de12138efa8441649b68cd086b650d16\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dacf66e9933a4bfca1ddbf155c18c4cf : SUCCESS in 13m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7e8c7235ea7f42d9b85c076a576c5f13 : SUCCESS in 1h 33m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae7a0674853c45dc93d89a5fa7cb0a0e : SUCCESS in 1h 48m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce0b7b96c7264041b0c17596ad42d91c : SUCCESS in 1h 33m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79a3e2821aa042e7b5d45e289d3f520a : SUCCESS in 1h 31m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd972eb7a086460ca43f8dc7afecad7c : FAILURE in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee89a6c7088a464c9ac20b1609474f9d : SUCCESS in 1h 40m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/244e0b3cd9c34a2ea1f007031a28d2fc : SUCCESS in 2h 24m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c6d679b7bae4a0894cd0897d1a52c81 : SUCCESS in 2h 09m 03s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97bcdefa6b024581af32a8f923dc1616 : SUCCESS in 2h 03m 38s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6d0aa3ad35d4974b71917c1961161aa : SUCCESS in 2h 06m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bb3f10beaba463f95c45687729d329b : SUCCESS in 1h 06m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8d305bec7bb49e48d7956580dc526b3 : SUCCESS in 1h 02m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09492734a1e04a28829c7f0bae7f6824 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"acb0daa69f72af86f8d9e1f1d7ecf78f8e078031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-20 16:10:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5426caf6b37810bf1bcfc44824a5f0eb3197f6e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 19:15:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d70e3d8fc0dd43ceb79733876c8ee4bb\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2520a3b340e74371851a7ccd74df2dd9 : FAILURE in 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c0060f4a99264fcf8255ceb03536c1bb : SUCCESS in 1h 11m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab9c45c22ad04ca2954b1fc5941f12d1 : SUCCESS in 1h 41m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/05f8fcf6f2d64f6aadb858877f8802fd : SUCCESS in 1h 47m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d404cfe2eaa44f96be91287bfe1981fe : SUCCESS in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/78e0a8ff29c746cd9a2147b3738bd7f6 : SUCCESS in 1h 39m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1b47c00666345d38d74cb4e308f1057 : SUCCESS in 1h 44m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37604b4f45eb4123b11bca54a95fb579 : SUCCESS in 2h 56m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9eded0222864589beaaad73b412bcd8 : SUCCESS in 2h 11m 06s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bcab9af6e26c408a8c7b6da984bb5746 : SUCCESS in 2h 18m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8172bb150ce646a582225a24b339f20f : SUCCESS in 2h 29m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ed64eabb7ac4103858a5ebdd5724cf6 : SUCCESS in 2h 20m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0817eb672704aa595824e45cd8948c9 : SUCCESS in 2h 10m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb4f4ec61b144ff99e90e95be1a862be : SUCCESS in 1h 10m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d2892a33a0a4274add753f6abcbbc0b : SUCCESS in 1h 03m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d3d730e52343f99b4b53f43e73a20c : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5b5abcab7a4ce6456adcec323b37cf6982645d61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-11-06 08:33:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"203fba28da7a0451507cb7ca6e1cd84b73a496b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 11:25:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb04487f810e4dedadad6658245d2d3f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/109eddc0fc9843e0a6d72739cfe90728 : SUCCESS in 11m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fec169e37f9946c9b453779fa82e871c : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ca0f77179cd4d54a6fb06413a6fed6d : SUCCESS in 1h 13m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/deb7ac1c19214cc98ec781d8fcb4707d : SUCCESS in 1h 39m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/757928beec804db9913734a9ba4293b4 : SUCCESS in 1h 41m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73cf0e1a9c724b62810d1417d2b6da29 : SUCCESS in 1h 17m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10a6f1ff36bd4d3dae1796a5c57872a3 : SUCCESS in 1h 35m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/573e2133e9c645fa9f49cdc73090755b : SUCCESS in 1h 28m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7547fadc425b47fda5f14b4ee10c5d18 : SUCCESS in 2h 29m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52b86bd819e14fdab511065967628c04 : SUCCESS in 1h 45m 35s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bc386fb6b5748caa83bf394e78fdea7 : SUCCESS in 2h 43m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d0226078ae94f1a88281b35cd8ba8e5 : SUCCESS in 1h 59m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/598e59d961184ad297a00e0e3d53192d : SUCCESS in 1h 56m 33s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea38dcf97ce64d8eb713530d92577d57 : SUCCESS in 2h 24m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49ed610dd3344cfda75158600797cfc8 : SUCCESS in 1h 04m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa489d2840fe4a08a21d336b058ce18f : SUCCESS in 57m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f452a42ff47463fb3f80014d9c36309 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"885c7ac08f4edb87dc7137133570f806740839bb","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-11-06 15:44:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"34d276086513ba96d7b75e9ba38eacb92268a080","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-11-09 15:26:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"86aeb159bf0d689b1961e26e6f98ad70e87a2468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:48:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a14880b8b10e85bf1dfb36e7ac784e7b68117ad8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 19:54:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dae5f3bd58ef4960bba06fc2a6df863e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acc1c4b05c154a6a98bb02f43f51f224 : SUCCESS in 17m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/658469de957e45a98a7f4d0ea67683fc : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/25f64617f89042b8ae1a8affb231b232 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17e7b7765fe94550b141e1f524210b17 : SUCCESS in 1h 17m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6352a1b12a374521af0289242b5f7dcb : SUCCESS in 1h 09m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d6cc912e23574707ae949e67b325b8c0 : SUCCESS in 1h 31m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f8ef69f70314904abaa46de69424da7 : SUCCESS in 1h 15m 51s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db685ae790dc4a04a596b119e2cf8cbb : SUCCESS in 1h 43m 36s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccff9c6b7e934cf4b1049f50a676f036 : SUCCESS in 2h 02m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26fa2e4ded774ee09e43a4d963dc00e6 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"777c7dba4124ec9a60bc8580bd89f71ab8154450","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 19:54:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d247601e6eb92c6011147cda9823eb41be76225e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 19:56:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1579b2298da5460dab597449bbc6c21b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b336c7f69bc495499e8a05f3cacb072 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"289ce991c420d7114b3a71a95602404e8ad7b0a0","revisions":{"08ca711fdae81881176669a90bd47873d01aa049":{"kind":"REWORK","_number":1,"created":"2023-08-23 10:16:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/892489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/892489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/1"}}},"commit":{"parents":[{"commit":"946c452af78aee58eb11ba4988e79994d10b739c","subject":"Define constraints file for docs and renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/946c452af78aee58eb11ba4988e79994d10b739c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-23 10:16:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-23 10:16:29.000000000","tz":120},"subject":"Use correct default for glance_mount_points","message":"Use correct default for glance_mount_points\n\nBack in Yoga we\u0027ve refactored way of mounting remote filesystems\nfor glance while keeping backwards compatibiltiy [1]\n\nHowever, there was a copy/paste error made while defining\nglance_mount_points with old format of glance_nfs_client.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/837550\n\nChange-Id: I755822089bf406f532a718db5b84df553cbc2084\nCloses-Bug: #2032771\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/08ca711fdae81881176669a90bd47873d01aa049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/08ca711fdae81881176669a90bd47873d01aa049"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"946c452af78aee58eb11ba4988e79994d10b739c","is_merged_in_target_branch":true,"change_id":"Ic65b9abc930e151f06d69fb9474a813cb504d7ed","change_number":891740,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"922757f95771b5e91d6d437607369d970aa28901":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-20 16:10:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/892489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/892489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/2"}}},"commit":{"parents":[{"commit":"8e36b597454b48228f7fe920909f8486b8c04bc2","subject":"Split lines to not exceed 160 characters limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8e36b597454b48228f7fe920909f8486b8c04bc2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-23 10:16:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-20 16:10:16.000000000","tz":0},"subject":"Use correct default for glance_mount_points","message":"Use correct default for glance_mount_points\n\nBack in Yoga we\u0027ve refactored way of mounting remote filesystems\nfor glance while keeping backwards compatibiltiy [1]\n\nHowever, there was a copy/paste error made while defining\nglance_mount_points with old format of glance_nfs_client.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/837550\n\nChange-Id: I755822089bf406f532a718db5b84df553cbc2084\nCloses-Bug: #2032771\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/922757f95771b5e91d6d437607369d970aa28901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/922757f95771b5e91d6d437607369d970aa28901"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8e36b597454b48228f7fe920909f8486b8c04bc2","is_merged_in_target_branch":true,"change_id":"I05029ad2a2a8e351ecc7236cb779a045c3a44403","change_number":895682,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase"},"289ce991c420d7114b3a71a95602404e8ad7b0a0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-06 08:33:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/89/892489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/892489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/892489/3"}}},"commit":{"parents":[{"commit":"2fcec5f55f3748b6caf0fa028cf2fd21264cdd21","subject":"Fix example playbook linters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2fcec5f55f3748b6caf0fa028cf2fd21264cdd21"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-23 10:16:29.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-11-06 08:33:09.000000000","tz":0},"subject":"Use correct default for glance_mount_points","message":"Use correct default for glance_mount_points\n\nBack in Yoga we\u0027ve refactored way of mounting remote filesystems\nfor glance while keeping backwards compatibiltiy [1]\n\nHowever, there was a copy/paste error made while defining\nglance_mount_points with old format of glance_nfs_client.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/837550\n\nChange-Id: I755822089bf406f532a718db5b84df553cbc2084\nCloses-Bug: #2032771\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/289ce991c420d7114b3a71a95602404e8ad7b0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/289ce991c420d7114b3a71a95602404e8ad7b0a0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2fcec5f55f3748b6caf0fa028cf2fd21264cdd21","is_merged_in_target_branch":true,"change_id":"I6962b83ec44db6d9edd0da6efbb58e1485f812a0","change_number":900155,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
