)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~900477","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2F2023.1~I755822089bf406f532a718db5b84df553cbc2084","project":"openstack/openstack-ansible-os_glance","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-13 10:57:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I755822089bf406f532a718db5b84df553cbc2084","subject":"Use correct default for glance_mount_points","status":"MERGED","created":"2023-11-09 20:43:45.000000000","updated":"2023-11-13 10:58:00.000000000","submitted":"2023-11-13 10:57:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900477","cherry_pick_of_change":892489,"cherry_pick_of_patch_set":3,"meta_rev_id":"09bdfaba41278d8b97da737c5bc59faf8e71e9ef","_number":900477,"virtual_id_number":900477,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-13 10:57:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-11-13 08:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-11-10 09:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-11-13 08:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 23:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 08:35:06.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"fd5354cebc22be4f04cf02e3aef1e7ee988a4ce2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-09 20:43:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"88d21297d7cc61b686074e8ddefe5135bd3d758c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:37:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ecdddb30d044ac3be54b9d3b2d2759a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f444fca5b80451bbf87f5cb8347a11e : SUCCESS in 11m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/503d814a1dbf49ad90f432426e615729 : SUCCESS in 1h 29m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/92dddd11c4c140e4b98cb307af6052aa : SUCCESS in 1h 21m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e5b7c516bc7462e972d7eee1ac55ab6 : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0063b16f63c743e7948036a9f6cfe35c : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/213b5760965545a09d62d48298e8f1a9 : SUCCESS in 1h 31m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a673020fa05b41d19bfea354357584a0 : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46e640952b904946b8c126bc18b521c0 : SUCCESS in 1h 33m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e15bb3b58ad4296ad982c4f4489e921 : DISK_FULL in 2h 15m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53347edd37e64fc0979d07ccd4da4092 : SUCCESS in 2h 18m 09s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5abf0fa32274be28f0c49addd91b11a : SUCCESS in 2h 03m 11s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1a58a9d18764d36b376b2fbda6c9e9a : SUCCESS in 1h 49m 14s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9b3e0bba0ef4cfb9dfbdd31ce8314fe : SUCCESS in 1h 59m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfbf888313b3489d8abf37db4e5438f1 : SUCCESS in 2h 00m 58s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57345daef44549e9861a8fc0572221b8 : SUCCESS in 1h 50m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f95cc2d27b74ce3b081696ec350974a : SUCCESS in 1h 06m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e674a5fb1912478283ec3abe970f1767 : SUCCESS in 1h 05m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff659b89c59d452dacfa687a53b6a882 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ccbc846b9c5150d6c24a1501d5295f822a7e9a7b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-10 09:14:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37a2c8a7f13a2044c5bbb3a0b9f78842572cfa85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:38:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b514a648f2cc40a0845df53089f03dd9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cec7cde66794446a73fa7efa8a4b750 : SUCCESS in 12m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d1c0acc197fb4d6f9d1aff7e4c124c94 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5189bdfab746432c831414b3dc4299b9 : SUCCESS in 1h 22m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4bef582c83ee4c6a832ac52cc7863719 : SUCCESS in 1h 32m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae059fedb5ee4fa8bd5fba91cf2035cc : SUCCESS in 1h 29m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce13cb63c86240fd99a07eb03a249a4d : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9465dda2040a4fc09831e46671fa8d23 : SUCCESS in 1h 41m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e4dc0692ed8434cbf3964703a97fcb3 : SUCCESS in 1h 24m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9c9a91c32e24331bd5435744b3c9f16 : DISK_FULL in 1h 59m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25a117b61c1045f5bf5625b7741508bc : SUCCESS in 1h 45m 21s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86fbac420230458f8548ec687eeeb176 : SUCCESS in 2h 13m 40s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3a930b1f3124c0b9277ff8fb83558a3 : SUCCESS in 2h 07m 27s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c22c13f1b5e478eae4c81574a291fe8 : SUCCESS in 1h 50m 41s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f9c37b2cb6142348d6bf8be0c641684 : SUCCESS in 2h 21m 05s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5820540d89440b6b3591d644d023646 : SUCCESS in 2h 10m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a37ca5dbaefb4e40a8a36c1ea2b7ed87 : SUCCESS in 56m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/162fb9008e8b42eaa00e1930690feabe : SUCCESS in 1h 04m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f051dcfd63c244b4b6a224c3bf9b628b : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a696b773879fca29fedc3b61374216690701cf7a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-10 16:57:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3fe6a21f99fce9c850072b8ab290f2914494f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 20:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9567f7871a14df9bbd537b6678944b2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db8cd599996c4ac4b4c042cff0c6b324 : SUCCESS in 17m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ec574c4800b943b08bc631eb330262f7 : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/338e6e5c334b426da32618713fb51dcf : SUCCESS in 1h 34m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f89591003c534383acf46cc395c5f558 : SUCCESS in 1h 37m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abaa5bf898b744b7877842aede4b9b2a : SUCCESS in 1h 53m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06683f009e2c4f11be85409c429d632f : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0af237a221f14fbc97b08d8436b0baf3 : SUCCESS in 1h 49m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/653d88b442d04352ad9cc275e7704c51 : SUCCESS in 1h 38m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/274a0133ae2a4567bc7b6367f8e1d2d3 : SUCCESS in 2h 20m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b07f60d095c24b3f982b60369e9dc2e9 : SUCCESS in 1h 51m 46s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d07342c1f7644278803c03535c48b64f : SUCCESS in 2h 04m 41s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1cab1c3c7cb4776bbb21647cb237896 : SUCCESS in 2h 04m 28s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ec80a1e73b2415bb41ce2302b250cdf : SUCCESS in 1h 53m 35s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4187180eefb143b69830871c033e922f : SUCCESS in 2h 30m 33s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4237944a536949b18721af9d7cbcb806 : SUCCESS in 2h 18m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb47fe1497a64a469dcc316bfe9cbdd8 : SUCCESS in 1h 04m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3c27237ad3c4e4ba4f3d11676db72e4 : SUCCESS in 1h 01m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef841ad4f914bb0bee9957a8e1d6a9b : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d0c3bfd83f0c8c077d149859d64a4a86f410d2d7","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-11-13 08:35:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9aac3cb374cc30fc38007599fd8f0b3f76ae98e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 08:37:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8bdf492e1091c9c6b478f8db4d801dc89207ffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 10:57:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/792b4abb639b4453b7ed0fbb35dcdba6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/923f4c4b8777481abddd763c85dd82af : SUCCESS in 11m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c6c4bfc42e2e4e7b810d830f69027bc1 : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8cb0821f67745f6aa1df9e73ab8e68c : SUCCESS in 1h 38m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb30540e01054b28b4f9af303c132bf2 : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/54b8ef44ca044727a9ab7e23170f9fd7 : SUCCESS in 1h 38m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/942e84e9db37456dbb8da9045e4f24ce : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3838ff3a47aa43f785a5cc3951b29926 : SUCCESS in 1h 34m 46s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eb973119f504f06bd2225e44e66fa15 : SUCCESS in 2h 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb5971594844d2bb722d4873da87873 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4cb2bbe8393e6ba32478cda5f95edbb6dde6eb02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 10:57:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"09bdfaba41278d8b97da737c5bc59faf8e71e9ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 10:58:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31c9bbbba7e84957a1036b69235eba7a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f9e3d5484548688856045756459524 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0b9c42a246928325ca7bce9c82c8c3735c8f37b","revisions":{"f0b9c42a246928325ca7bce9c82c8c3735c8f37b":{"kind":"REWORK","_number":1,"created":"2023-11-09 20:43:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/77/900477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/77/900477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/900477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/900477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/900477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/900477/1"}}},"commit":{"parents":[{"commit":"32f729b3674c2de9c5e3adaa9e09f3ac1e6f0c56","subject":"Use proper galera port in configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/32f729b3674c2de9c5e3adaa9e09f3ac1e6f0c56"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-23 10:16:29.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-09 20:43:45.000000000","tz":0},"subject":"Use correct default for glance_mount_points","message":"Use correct default for glance_mount_points\n\nBack in Yoga we\u0027ve refactored way of mounting remote filesystems\nfor glance while keeping backwards compatibiltiy [1]\n\nHowever, there was a copy/paste error made while defining\nglance_mount_points with old format of glance_nfs_client.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_glance/+/837550\n\nChange-Id: I755822089bf406f532a718db5b84df553cbc2084\nCloses-Bug: #2032771\n(cherry picked from commit 289ce991c420d7114b3a71a95602404e8ad7b0a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f0b9c42a246928325ca7bce9c82c8c3735c8f37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f0b9c42a246928325ca7bce9c82c8c3735c8f37b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
