)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~901189","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Icdae86c7aa6138a5541a335f929ab9a90723b194","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-06-01 12:19:25.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-06-01 07:03:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Icdae86c7aa6138a5541a335f929ab9a90723b194","subject":"Add test scenario for cinder as a storage destination","status":"MERGED","created":"2023-11-16 17:12:28.000000000","updated":"2024-06-01 12:21:27.000000000","submitted":"2024-06-01 12:19:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901189","meta_rev_id":"be74db9389ea414ba8114f033fd31635fc763e26","_number":901189,"virtual_id_number":901189,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-01 12:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-20 10:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-20 10:49:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2024-05-20 10:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-05-20 10:49:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 19:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-25 14:11:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-04-26 07:22:15.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-04-30 01:46:31.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"3622ce25f58bca7828d39d3b892f2922976e6d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-16 17:12:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e16b4b28a20b147f5b4eb148308920c999a7082e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:49:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e8c679ea6db49b9bf150d1f88bb7f81\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea9d4a6acfcc4021ae61bdf5d374a6a4 : SUCCESS in 16m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ef991b547c8046ebb1c23c31e3cea1e1 : RETRY_LIMIT in 14m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/34f4a2ea81b146659987918a608a33e3 : RETRY_LIMIT in 15m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/845ffa3afbf24ce29e2bd2cb55ed2229 : RETRY_LIMIT in 17m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3216114861a54134b7d86144f1011ab2 : RETRY_LIMIT in 14m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbf93a45b0d7472693f02e3de9b036a4 : RETRY_LIMIT in 14m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/72edbc170dcb4b6fbcd560e544257e64 : RETRY_LIMIT in 15m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39a61a5f87b34816b5d9f61149dd4656 : RETRY_LIMIT in 14m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5c5e00f49f844ceae5e0b71aa555975 : SUCCESS in 2h 28m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/768f35e0e2a944a1b45c2f6de6cbaa05 : SUCCESS in 2h 24m 30s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75a15317eb2a4dc3bf23987305d749b9 : FAILURE in 1h 16m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a26b0b4d6cb546c789d68147d84e8d30 : FAILURE in 1h 02m 38s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b178ff93e85844cca540966d99cefa1d : RETRY_LIMIT in 13m 21s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b57ac1d6eea4f2eb4670d7be48cbbc3 : RETRY_LIMIT in 19m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a5d995cbaf945e5be7181b84a0b6ea5 : RETRY_LIMIT in 15m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5acccf664d0c43ad863e7991bf86e16a : RETRY_LIMIT in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c6d1dcab5d4c83a63bb7ce689c40d0 : SUCCESS in 5m 11s\n- openstack-ansible-deploy-aio_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6df2fcccb5b4d7c9d2a665109249d37 : RETRY_LIMIT in 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9ab152f2af3717bade3b3e240d53ecf8ebcd6585","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-21 12:23:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3999cde51bb8bf55f4abadc28fd4be289c5813ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 15:04:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13ae13546a24481bab5591feace8fa1c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1601644798ac470fbd154cbe6bd8e8a2 : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ef77f1ff4976405b823df3e00d1b54c0 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/99d7a18f130b481fbca6b4b4bcc6a167 : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8e2fe2a6852435fabfc58bb2f961b3a : SUCCESS in 1h 25m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f2756f6348f48ec91e52f09e1f33ecc : SUCCESS in 1h 27m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d7e961c5fae4ddc843ae4eb4a8332bd : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0f48e457bc04136a29c25544fdf2419 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6871958637e8464f98b2896d46dd80b4 : SUCCESS in 1h 11m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f74398c21684ff88c2e0bc41bf88d3b : SUCCESS in 2h 32m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/767260531abf41e08be23928fe3b1b5a : SUCCESS in 2h 18m 24s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55e56ca8abb3474583ad4fd5b7a23c9b : SUCCESS in 2h 05m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a044d132bc94b8d83226207e2fe93af : SUCCESS in 2h 11m 03s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de92e5db330d4b5a9f70e7450751ddbc : SUCCESS in 1h 38m 30s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c08dd88e4724605bfb02b145711b776 : SUCCESS in 2h 09m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/38f045d7865c4a238a678316caeb0b73 : SUCCESS in 1h 01m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/325b5e9a9fbc43558964af4e459daeb6 : SUCCESS in 59m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd936542aab549faaad0696c7b61979e : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f873570b37f4a57a7e667fd2991b8a5 : FAILURE in 1h 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dde2f922c930270b95cf0e0e64d3b9e75e244bfb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-22 11:53:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67c129fca8208f4a4e7003e84ddc995e4af55d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:34:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c53ab574bc03470c93a28915d06e1738\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea367231c0d34589b7b1d98b1133509f : SUCCESS in 12m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5e9c37ed267548629b1f7299543e51ff : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0fa8349e30e149249ab0300db24385af : SUCCESS in 1h 16m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3aa99cb321034f6a867b0b5a6888a1c7 : SUCCESS in 1h 12m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10a8734e6540485d83106542411e80c8 : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/140a8fd1fdf548d0b6eec7bffd8c43c7 : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fadc732e01db42818b4930063c377733 : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c0ffca2b18b4035b40957354b30d310 : SUCCESS in 1h 14m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de9d4b325c38490dbd49439b06e53cfa : SUCCESS in 2h 18m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/918193833eb44574a982833ceecb4e95 : SUCCESS in 2h 20m 11s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/959cbd15ef754955ae04a603d9fbed77 : SUCCESS in 2h 32m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5efd71cff8a48ad85b171388be369bc : SUCCESS in 1h 52m 40s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a41ca1a272d74fae841d4c9ba2a64b34 : SUCCESS in 1h 53m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4aaacda350e462a908099f80c9a6791 : SUCCESS in 2h 08m 47s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7294d8460b54dbb8748830645927d6d : SUCCESS in 1h 06m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3092282db454ccaabe8213e6211407c : SUCCESS in 58m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125ccfba42884d9e8a706099f8e17353 : SUCCESS in 5m 45s\n- openstack-ansible-deploy-aio_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/699ea8ce1cd94babb6b51969c51b7b67 : FAILURE in 1h 49m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f29d125041285b5b3b2768309f043388c40f7ac3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-27 10:57:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcca2778f411ee520a7ac3779b4f998d6a46afdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-27 11:37:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"adf9d53feca05532e8f81ea71f54b1ac5caa4902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 14:09:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48bbfad529304bbd949cc8ac423149fc\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4cf8338b7f34122a33abb5e29d4bb08 : SUCCESS in 11m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5e91f3efad314ff0a7c4d1499451a5eb : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0714ac31a5824ed0915b196517cd563a : SUCCESS in 1h 12m 49s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa86e6235d2d49afb7e7b1f8c1ae9122 : SUCCESS in 1h 22m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/143b7c38205d405ba85ffd0e5c0951cb : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e0fdbdde85c4520a5e8ab85f21a98c5 : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6835db46708247e4ba17518038803dd6 : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87d40ceb964e45eab0355e92c3b98794 : SUCCESS in 1h 11m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6cbc76da0dcf416396bebda0880f5b0d : SUCCESS in 2h 28m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5902887b3ec4a71a3ccabcaba7b1e32 : SUCCESS in 1h 57m 41s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96d9bb20f2f14c5cab94e3be5b9da41d : SUCCESS in 2h 12m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/143f7a7cb9d64d908739d4c0fb2244f3 : SUCCESS in 1h 45m 52s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2025db9a55545d28f3674d8d784bc8b : SUCCESS in 1h 52m 29s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79c72d48a0594f018d306cca0caf3ad0 : SUCCESS in 1h 48m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81d6b457229e4f9380a9ff98697ca91b : SUCCESS in 1h 04m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa569eed946b498286f0000a2a4dc231 : SUCCESS in 56m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53726d5e47040fab0d558ff5b06c318 : SUCCESS in 4m 52s\n- openstack-ansible-deploy-aio_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e523a0a8ff4a40ff83b93a4960e95a51 : FAILURE in 1h 54m 50s","accounts_in_message":[],"_revision_number":2},{"id":"aa2e597c2197bc251271fd532ca9772da8cfcc71","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-28 15:17:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"438756e52e68c7de277584810d238d4589c8b454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 17:39:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cad705f8c5a493196dc87f9d0fdf1ec\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9aad81e2abff4e8ba0445407a5ed76ca : SUCCESS in 11m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e7d0c50779924104946060814fb9d0cb : SUCCESS in 1h 12m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a3ed9bcd33db451cb8c1ef7e962651bc : SUCCESS in 1h 14m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bbdf17d205d4b9f9fde6404294bc2b3 : SUCCESS in 1h 19m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa8ea451dc984f68a0286085ec611405 : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a48a4011f6a34a18b1574f17a8fe0d51 : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6a978f6b92d74be8805c06931b640866 : SUCCESS in 1h 24m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/934f490277fe4dfeaff1bc1f2914eedb : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67794f678329427cb6df8a2f920e88a1 : FAILURE in 1h 09m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18e4e2053caa4bec92a87fa31232ec88 : SUCCESS in 2h 05m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64d8151f562f4b84badc0334fb666a03 : SUCCESS in 2h 11m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe400a181216445aaf14e0f0e254595f : SUCCESS in 1h 39m 25s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf4c6c24040c4551be5bfd4e0c485360 : SUCCESS in 2h 16m 53s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c6e04ea145bb4ba988c2595af6fe3560 : SUCCESS in 2h 17m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08f44cdd77484fe7b827897dbae2f517 : SUCCESS in 1h 40m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/621b794c713b459094a8a26bd9232697 : SUCCESS in 2h 06m 11s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f376323e882d41eb91383b7c3cda7022 : SUCCESS in 1h 55m 53s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15bca735568b401bb8714738e71c4d19 : SUCCESS in 1h 05m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c31ebe3b9dd240b9b46c806dcc4bf047 : SUCCESS in 1h 01m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04de46ddae846238f8bc728b46c4497 : SUCCESS in 4m 28s\n- openstack-ansible-deploy-aio_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a387a96116e430f8eaa090543184032 : FAILURE in 1h 41m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b1c5f88f7dbc25afc77978618511002720afa750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 11:05:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d4518c95283ba9473532f20ffb5816457cb746af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 11:05:54.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"81d1160bdadc4e084377503b03ea464c99524b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 13:30:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a1fa159beb541748b136439df542908\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2f77490148942ad9495fd3eef63d1ce : SUCCESS in 12m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/569a485f34ac441aa47614cad75f8152 : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/26a8968135f844fa82829cacbcf21b43 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce75d49618914e63a674cf3958413d02 : FAILURE in 1h 10m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ba4cc730a5c467aae6d8735bad412a2 : FAILURE in 1h 15m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67cbae2fc91c47cfa55c2d74ac7eb205 : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/255c38d93339478fa285216311e449f6 : FAILURE in 1h 00m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46b17fb99a6d49fdaa2c73eeae0ade89 : SUCCESS in 1h 17m 00s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0c688f70e30421da5ad3043f4b01fd4 : SUCCESS in 1h 15m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59da468701f142e89d00726d48c9d13f : FAILURE in 1h 40m 21s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d9a75edc59e455eb14dee7b2b6ca3d0 : FAILURE in 2h 12m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8eabc64e885d4a06af241d9a0a2241c2 : SUCCESS in 1h 37m 25s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ee06f0dd1b848929a149f3886bc8319 : SUCCESS in 2h 20m 33s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b7679b5cd534a23abfac2ee877d1067 : FAILURE in 2h 00m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3116d279cf174699b22d1380eb4d744b : SUCCESS in 2h 00m 55s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3091f57891664bb78363650c857b37e8 : SUCCESS in 2h 05m 00s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03883f046dd445f08fb3f2bb547c5c5e : SUCCESS in 2h 14m 25s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7cc7a4a5c514485ba59188a3c940781d : FAILURE in 53m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e99745483874f00834a4c72979d2519 : SUCCESS in 1h 02m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8fc93b7eb545c2afec26d01b58c097 : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/757f078df8b6402797b7318c23d087cf : FAILURE in 1h 38m 54s","accounts_in_message":[],"_revision_number":4},{"id":"d8d3644ce080cbcdc2de40f3372170fcf2021359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-03 11:54:18.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"96baa5ed31e068f4b06b85e599c5e39cfa79052c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-03 12:05:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"87ed2a0a85da014f7f15ddeb426067f58e06e96c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 16:39:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/545544eb10104634a015ca2890e26a6b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b11ae0ae53de4962a57183f12ee29993 : SUCCESS in 11m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7f9bcb3589f846f095133f7bcdf4df5a : SUCCESS in 1h 01m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6ddbdbf408504dd6985a779e625065b7 : SUCCESS in 1h 13m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8715937a46b848dc824a6ee1f6cecbc0 : SUCCESS in 1h 12m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9f30556251c4b32b067ecb0c78dee43 : SUCCESS in 1h 31m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e352a74d41584fc6a3fb56ca94470402 : SUCCESS in 1h 12m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df34da8783d44b3f9b9a7afa843c005c : SUCCESS in 1h 30m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0172f04d6f94d5fb36074d221ef058d : SUCCESS in 1h 14m 16s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdcc061e5a664719aef0e92707981e3b : SUCCESS in 1h 33m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f85a35ae7d44a22bbf86d8570d1450b : SUCCESS in 2h 02m 38s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9bb187b9348b46b29a1b74fea84c65bc : SUCCESS in 2h 23m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f0541ee72e1472d8a8bd3a0bef897b1 : SUCCESS in 2h 04m 19s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c88aa27d8c04299b4e8681163c7ebcc : SUCCESS in 2h 15m 42s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bd622d820f346c6a676dad8f12b5340 : SUCCESS in 2h 10m 03s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e1cd0436ca247ce9b7463ceb2a97058 : SUCCESS in 1h 47m 15s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/323ca2b4831a464797f92cf67975e02d : SUCCESS in 2h 09m 59s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54304545fb914c579bf8396e6b0c1252 : SUCCESS in 2h 23m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ebbd0f51a66d406d8a64a03627c71e94 : SUCCESS in 1h 28m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72f5fc97cc954be1bca8d22b2ac1be16 : SUCCESS in 59m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b158f749332d444695bf80228cfa3b49 : SUCCESS in 4m 38s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/564e76de52e6439a80e9b0910fa94387 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":6},{"id":"e5dba0ebd6372153289b8aa1aba6e39d85ab8a07","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-04-25 14:11:00.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"43728766035b2496c9028cb299561d21f494d67e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-04-26 07:22:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"801f105cc39f855fc7d105195bf49250b31d2c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 07:24:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"90bc9b1f6017d63efb7eed15a9b483d62c214877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 09:59:34.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04ac099788dd406bb334b3126b4e06ee\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4a5f0f675924fa5b6f9d26ef2323fb6 : SUCCESS in 11m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/90f43429cd854e1080bcc247bea30755 : SUCCESS in 1h 10m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/51550530ee5a4d05a3a2b4df99a3a211 : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21e9a2e9d6b44ae488be6b1c92b3728c : FAILURE in 1h 05m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/835a59a8f1634ce182867da36b8a4b38 : SUCCESS in 1h 13m 07s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff5b6c6459664c6ba80b01fdcfc6a604 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17401284e61d444c86813c71e9947388 : FAILURE in 1h 00m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/884b56d2bb6c428bbc6766c9830562b0 : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c15de4145b5049fa820ca34add79456f : SUCCESS in 2h 11m 06s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9133cf42dc26462bade4aa59ca82973a : SUCCESS in 2h 25m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91550e6a1b044b5e9741076f18412c4c : SUCCESS in 4m 28s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50d01769c7c94479a423be18db9e39b8 : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":6},{"id":"a237ca301740bae751ab29f45c5687360f347ceb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-29 07:39:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6f6a2d3383625f6751e8fc2c195ffc9ce3f440bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:09:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/248cf3851977489aae18633d27f44984\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d285315beb9149379ac151ed7db81d8f : SUCCESS in 12m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/53493f051ea140088bee8d6978fb2008 : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9cc093535fc74072b26f5c678231bbfd : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/774f51af4b434812a96b9620ebe96300 : SUCCESS in 1h 19m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/276d38d6be094bb39c2b3b5013609e6a : FAILURE in 1h 06m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7028f71064cb4262aeedfc48ce7801d0 : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c9a8420443344c0b5a049b2ec8bd837 : FAILURE in 1h 19m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2b8e5db1e374323b84ba71ed7243c95 : SUCCESS in 1h 29m 35s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bd0bcd9370f4a66b78e30da73a6c753 : SUCCESS in 1h 17m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9bae4518d5394c7abf4497eff2a78671 : FAILURE in 1h 05m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4259361272224a7786bedb51acfdcecf : FAILURE in 1h 01m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef3abade19144f54b6de090a5a2ed57b : SUCCESS in 1h 42m 32s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1846872d8d264c25be4b7c0f627a502c : SUCCESS in 2h 07m 56s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8c7fe1ded67d481495914d33eb76c997 : FAILURE in 1h 03m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3434334b19f847299c5c319f712e0be7 : SUCCESS in 1h 46m 10s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25b9b8d1c00144f7b9cb83ab4a4a7919 : SUCCESS in 2h 04m 21s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6621a7784e5461d967a6015511f1bab : SUCCESS in 2h 16m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/05cd3b8601d141c0b5348681548e34cb : FAILURE in 52m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/827e56a974f54c248b35428ea1531c79 : SUCCESS in 57m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfcd3d03b4404ef38d6ddc9d8fb3f9c4 : SUCCESS in 4m 03s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1245edf3cf54d3083e05c7b56bf8d14 : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":6},{"id":"ea2497b4f36b33e25f01495bd35428a8dc8b44c0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-29 19:15:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"125f644630309ed07742eebbb6423d0952e80273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 22:03:22.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8de296b1563f43a4a7ed9c4f74761290\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61efefe8c6244632af6be4beab9283ce : SUCCESS in 13m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a0b3deaf09fb46ae92df4cf034907db9 : SUCCESS in 1h 10m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b0da3e7568284a949bc41b3ca6a95aa8 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d855c59ebe5d4724b552d6055b4edab6 : SUCCESS in 1h 22m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d9779f492ef49e1bb201090c30e99d0 : FAILURE in 1h 19m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0e8c24caa284b6b9fc7ae77eacad7c0 : SUCCESS in 1h 23m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aca082c59ceb4d7287b9f0dd03f1a606 : FAILURE in 34m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/582df3cf2b4641b69d13bb5c915f3d0a : FAILURE in 34m 31s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb8c831238e04a5fb5e7443b4b4944b3 : SUCCESS in 1h 20m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/09465f0accef4b24aa69a726856be7d5 : SUCCESS in 2h 11m 56s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/413133fb3a7c4546b27fd083e294f83a : SUCCESS in 2h 37m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19d86caa28f94c9193e05dd912f22c07 : SUCCESS in 1h 41m 03s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22e874c0d468462fab5877c64fdc2c92 : SUCCESS in 2h 18m 36s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/193ce06d52c94681ab11b83569c6438d : FAILURE in 1h 15m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be6b894e18ba43e89d867f005ed570a7 : FAILURE in 1h 15m 27s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/485ea29ef463448fa8863d6787336583 : SUCCESS in 2h 08m 13s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ff624bfe43a4241a4fda1446e630545 : SUCCESS in 2h 20m 39s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8facb29865f46b096ac9669b890c7e3 : SUCCESS in 1h 02m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23642657174146f2bf198c50be816316 : SUCCESS in 1h 04m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05ee3234c164a93b6dafbe624924e80 : SUCCESS in 5m 16s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4265b29dafc4ccca3538311a42e2c02 : SUCCESS in 1h 17m 13s","accounts_in_message":[],"_revision_number":6},{"id":"3a8cfca4f892c2f4a9f2e99721e180bcbcaa51d8","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-04-30 01:46:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6e182e4d9e25e30690d192f30efc06709d599f4c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 08:56:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ea97fc646ba8acc7934771d95fa242f83d909d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:54:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fae7f4c432a4dda8a22770a2355f1e1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4394d2d07aab4b7aace2c9108161c5ec : SUCCESS in 21m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/67e55a8036c243fab3cac22724c1e09f : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9a40d6af6cc748b4b0d04c3e40a2d2d0 : FAILURE in 1h 25m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b053876a1ad84a98ab5221244a5a79fc : SUCCESS in 1h 24m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d992c1a6e28a4f77905fa2173de0ef88 : SUCCESS in 1h 33m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dabc0c108cc64ffdaaa4b625f01542f2 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7d9c915325145329d21a3cf499740c9 : SUCCESS in 1h 39m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3843660fdc494947b7f8a47f26254ce7 : SUCCESS in 1h 26m 06s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f05b2a6a24a4f1bb6dda1caa17d3bd6 : SUCCESS in 1h 27m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d8f1fd029b174ba9b2d88fe459df12a3 : SUCCESS in 2h 03m 20s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa3cdc91001c4215aaaee743c834c1cd : SUCCESS in 2h 50m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc44f14479f7447181edea6ce6afb6eb : SUCCESS in 1h 58m 56s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06b7012ae01843ba9c05ca5ac625730f : SUCCESS in 2h 16m 26s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/66b35ca603344d35aed9d06690b96afc : SUCCESS in 2h 14m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3f1dc4aa0647559c6ee94d1ccf1794 : SUCCESS in 1h 57m 14s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd5ca96cbd60440ebec6e7be4e2a1ec5 : SUCCESS in 2h 11m 56s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d249a79a4c94476b45a34fdc27f5d3d : SUCCESS in 2h 40m 49s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87a67f255aaa481abd04780855f0d9b7 : SUCCESS in 1h 08m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a9658b8235d4079801b5dade27db6f5 : SUCCESS in 1h 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85dd76e04cbd42a4871409ebb3d3ef79 : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/510113dd7d2d44dfb1571f3a13547cec : SUCCESS in 1h 33m 55s","accounts_in_message":[],"_revision_number":6},{"id":"246fde323805fb9ef732c02e68b711796fea972e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 16:10:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"eb73b1c0d9dc5eb05b72c060a16cc19eca08cda8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:47:41.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d814e24ac114a158efa39559ea45869\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d86588a3aa44e7ea5b7a774d63be5fc : SUCCESS in 14m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9c31c7198bae485bb0cc881028cddf87 : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6a39600a6eb141209702af47f5331eb8 : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6a69b48d0684799a6fc339800890b81 : SUCCESS in 1h 24m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a8b5c51adaf444019996cf8778cfc1c4 : SUCCESS in 1h 28m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d37244df78924e06895d846e31c29dce : SUCCESS in 1h 17m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8e580133ca5449abd17294b4c375c1c : SUCCESS in 1h 39m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cfe0dad1f16480daf8412855fdfc863 : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7724b9add9354b618f6550a78e1a9eda : SUCCESS in 1h 18m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/efbb63f9157b47f6bd4c0e4589aaef48 : SUCCESS in 2h 13m 15s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1642affa9c954c5081dfea3033b23ce8 : SUCCESS in 2h 30m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08a14e33f57b45bda3d08deb5b174016 : SUCCESS in 2h 01m 22s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/252ee35772c24f5aab2c04e08d818523 : FAILURE in 2h 20m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b2053b209d5448786e4a00387e3a296 : SUCCESS in 2h 15m 52s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c1cbe7789aa452989546c0d36f3b8c4 : SUCCESS in 1h 50m 18s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1182b6064cee4d59b99874db3b63b92b : SUCCESS in 2h 16m 24s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7657925dff7347279c29225eb66bda03 : SUCCESS in 2h 13m 17s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e850222acc37454796552e91b813f3d8 : SUCCESS in 1h 05m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70fe474df72e4c4786f308b9015271f1 : SUCCESS in 1h 02m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b7acb5e2b4417d9ae9b098eb673f29 : SUCCESS in 3m 53s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4ebdf87afaf4f148f2e08e287d57502 : SUCCESS in 1h 29m 16s","accounts_in_message":[],"_revision_number":6},{"id":"dc13aa3de838641cc53ea44543794b02eb35b550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-20 10:49:36.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c9b23f6b89aa10fe423fe207fc6d2ff9790bb7ed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-20 10:49:43.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e76d1e8f77c8e0616076fb78ba1d9be118bdab7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 13:39:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/162ac2fc7aae43b181c83d7653e7b9c9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9e75fb8c7714677b305cccdf4d850be : SUCCESS in 21m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/97fec60559574920bc59aa85d008daa4 : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7d620b45548147489d25064cfc4b009c : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e59d481ba4ef4ffd9f9b514890348a5e : SUCCESS in 1h 22m 19s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d908e1b7caa94dbd8c36500d28fd1e08 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1d3451bd2354fea9569e05e2ad98b9f : SUCCESS in 1h 15m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6c5f632bb524f6985af13218eeda611 : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/643cf4999ece49d48adb628f61b480da : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/047a580b86074c8fb0c1773642dd7823 : SUCCESS in 1h 27m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/24a99c370bf5473096c3ce721edeee93 : SUCCESS in 1h 33m 16s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/03221db0fb44431792648d3d11b3ced4 : SUCCESS in 2h 39m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8414129c17ff4fc6869be3940328b7d9 : SUCCESS in 1h 53m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56a7c87ee5ba4f8388c19e59160a8471 : FAILURE in 2h 04m 12s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f83165959a7444cb623ef26edfaa15d : SUCCESS in 1h 44m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc27e73972af47ab87c3157e6d215989 : SUCCESS in 2h 08m 52s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f5d1a3af4724719becb958556fb9f8c : SUCCESS in 1h 54m 52s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30daef0f99294fa7a266e6e86e7af651 : SUCCESS in 2h 22m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6fc84390b5f64d6a8ca9a88b0d98cd2f : FAILURE in 43m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51bdecb2a9284942b6e01e64a337ef17 : SUCCESS in 55m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0997dd61be749b49b95adaec5444fa1 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa32c7a8ad2748f89524d2871cac0992 : SUCCESS in 1h 20m 57s","accounts_in_message":[],"_revision_number":7},{"id":"ce61c85e819cc03fc506d1cf980e1539523958d8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-31 07:20:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a87e32d169ee32b9987eaa0b676d511e2d5d2a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 10:02:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87c9d23c354a49ca8f4b7be6a40820b8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7e017d4201546b898cc2043671bc829 : SUCCESS in 17m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d03718412a804f30899c2c38d46f7d01 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee8c4b13ab9d4d1e9ef79e9fa6277ae9 : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f321bcfe86a04b1cb11da490cbedb7ab : SUCCESS in 1h 23m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/433d5a8afd074057b76f61b0f0289466 : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fecf829072a43ef8ce309a0e4b0bd02 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/adbde572d94a41af8f8e2c3b24505a3c : SUCCESS in 1h 31m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c63802d13734c0da97382dbaa0f0e42 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ff3137d04f74035808291ca49df7388 : SUCCESS in 1h 15m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0ad9a195741142738a416840c915dc57 : SUCCESS in 2h 06m 13s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3db9932e49944dc0a7e3d6997bb841fc : SUCCESS in 2h 35m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f54622d68a545088d7fa0b84b1b558e : SUCCESS in 1h 57m 30s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/769395002c6f465592d4c85b5fdd6409 : SUCCESS in 2h 21m 15s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55f40fb458f4482bb7e37719f42044e1 : SUCCESS in 2h 09m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dd3eb7ec73843e3b3425971ade4c0c9 : SUCCESS in 2h 00m 25s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6637472195404addb7a88e12c8112812 : SUCCESS in 2h 05m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4018bf829f6d457e983009679e10d1fb : SUCCESS in 2h 14m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a98e5da6cf89403d965ff4baa0dabd7d : FAILURE in 53m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11a2d06d36534a1eb4423813c73e0beb : SUCCESS in 1h 00m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8c70ed70fb496ea97eb0a24445f4d5 : SUCCESS in 6m 05s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55876916e62e42c2a3b87f6fb0b01778 : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":7},{"id":"350e31ce65f8d18f496049286b081d8cac73ae02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 10:04:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3cf5dda2e1155bea8fd93c20ddfac9f445312a48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 12:15:59.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47f5b5b7f58e4106a5e59b32ee1c320e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4546e0bafe8f4084aaaacb546acbb1b5 : SUCCESS in 11m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fb7f00a5d54a4243af4412de4fc6d0ba : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ab138487ced499888206bbd88710606 : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/985964dd9bec4e4ab31a7f0b62b45875 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d42a7d0317cb4c4289bd90a01e6bbaf3 : SUCCESS in 1h 19m 26s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e13adf0057764539886a917a9aaebedd : SUCCESS in 1h 19m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/95b869d709404fd787b3ca047cd7e91b : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63cd01778387425c9258796391f4df20 : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4f4ce2c2441409b9c4f499e3a763ee9 : SUCCESS in 2h 09m 58s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/564ede169f6945a18f6deeb192fe60da : FAILURE in 1h 43m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6eaf1e5715e41df92b8e143bff67379 : SUCCESS in 4m 24s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66bd414f8d2644439ba5fa78851cb7ba : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":7},{"id":"dd230870cfa501fa950a03501ff4fea81ba33aa1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-01 07:03:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d4e3960473c2646005d0eb6c3d23140d765f400f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 09:47:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51bee9442e524ce0bac62277195d840c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72055caeeb1a44cdb669af3ca81e52fe : SUCCESS in 17m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/25371b7d23d048e4bccbdebc31613b3c : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ae16637111b44a0a050e8e426abf0a7 : SUCCESS in 1h 37m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a38777b24bba42cfac971807637d8640 : SUCCESS in 1h 30m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/72386c12602140e1a3db7bba93c979a1 : SUCCESS in 1h 27m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b72c998dbc1e4fad8de860a168e825c0 : SUCCESS in 1h 22m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4c96d1ad6be4feeb7f72989c62dcf6f : SUCCESS in 1h 36m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a9d492e6bc44f06a17e307b37ead1b5 : SUCCESS in 1h 06m 07s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98f1a983c9e9499c87cacba7ccbabc0c : SUCCESS in 1h 08m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c56cb11d895c4b2f923a413cff6e5e55 : SUCCESS in 1h 52m 00s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e37fce1a44a84404be9e7b53c002b11a : SUCCESS in 2h 36m 23s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49c770fcf98641d2a3fab28159f94490 : SUCCESS in 1h 44m 38s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/329d383b607d4806ad77db8ea8cbda92 : SUCCESS in 2h 19m 37s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c02a72208a7047948f52e0d3bc9f3d4a : SUCCESS in 2h 12m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1354e8172424569a00a08feb09fd14d : SUCCESS in 1h 30m 02s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4df2b7cc0d62465eb8658909d01719c8 : SUCCESS in 2h 11m 38s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9b5dd4877b84a45827ab8464b9a816b : SUCCESS in 2h 10m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a537e9aa20be46da952facb34773b743 : FAILURE in 50m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/375992745491431a982910657a7857db : SUCCESS in 53m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2008c114c4bf43878ac88d9114a146c3 : SUCCESS in 4m 07s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f140d7032eb9488b84eff94e00961570 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":7},{"id":"f344aafabc9d7f1abc35a95b4f1f60501167c1c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 09:49:24.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c686936ec786836096bf7a6c3f751ea893798bfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 12:19:25.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b5dc890ee994b139defd7a7feace82b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e93d43f174ae410c98ce7e185e9872a5 : SUCCESS in 11m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a42a1b58ae164484b9bda724d5099bf7 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c00789f8389e426ea0296f8c656c8650 : SUCCESS in 1h 20m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2f5b47922f940c1bd3262005153b31b : SUCCESS in 1h 34m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c30bc010b2ca4773b71e13acb14aea68 : SUCCESS in 1h 16m 20s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/673dc820e5234fc7891600bc6816e766 : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f1a7126445542b6a67351e2a5ccf354 : SUCCESS in 1h 26m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94b209e7b25a436c8b1793251db1778c : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e12386655e2d40eabc3cb5683b371f0a : SUCCESS in 2h 15m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52d725e46a054f73b5f4a32022f4fd64 : SUCCESS in 2h 19m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e4ccc96b734e988e0864e3ff4aab7d : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d966ed3a0b704c3b8ed970855a6d13f9 : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":7},{"id":"124f283d85c3a28ab49c45e7425b5c876f77b5d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 12:19:25.000000000","message":"Change has been successfully merged\n\n6 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":7},{"id":"be74db9389ea414ba8114f033fd31635fc763e26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 12:21:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c54d895f8b254f72a4df2476fef3a599\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41282751a054a44bcd260b67f4eda26 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f9b03dcb7681c1175e276052abf6b18af8aa4749","revisions":{"9421a83a7a273f64a4f776d20f427492055e3715":{"kind":"REWORK","_number":1,"created":"2023-11-16 17:12:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/901189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/901189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/1"}}},"commit":{"parents":[{"commit":"c6d37f6db798b509b44656618e4369fcc862c443","subject":"Fix cinder store configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c6d37f6db798b509b44656618e4369fcc862c443"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"subject":"Add test scenario for cinder as a storage destination","message":"Add test scenario for cinder as a storage destination\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/901187\nChange-Id: Icdae86c7aa6138a5541a335f929ab9a90723b194\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/9421a83a7a273f64a4f776d20f427492055e3715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/9421a83a7a273f64a4f776d20f427492055e3715"}]},"branch":"refs/heads/master"},"da0586c84f74280968ab4f1a3dedda8d2d683d8f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-27 11:37:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/901189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/901189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/2"}}},"commit":{"parents":[{"commit":"e4acad2bb660d2776bc86e30e7791f9f9adb5c4e","subject":"Fix cinder store configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e4acad2bb660d2776bc86e30e7791f9f9adb5c4e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-27 11:37:20.000000000","tz":0},"subject":"Add test scenario for cinder as a storage destination","message":"Add test scenario for cinder as a storage destination\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/901187\nChange-Id: Icdae86c7aa6138a5541a335f929ab9a90723b194\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/da0586c84f74280968ab4f1a3dedda8d2d683d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/da0586c84f74280968ab4f1a3dedda8d2d683d8f"}]},"branch":"refs/heads/master","description":"Rebase"},"19b4f673b7e832f4c9c18cf22b9c74b02330cf44":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-29 11:05:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/901189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/901189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/3"}}},"commit":{"parents":[{"commit":"bf25a670ff4a229d7412e85051577f7829168495","subject":"Add worker_self_reference_url to glance configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bf25a670ff4a229d7412e85051577f7829168495"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-29 11:05:39.000000000","tz":0},"subject":"Add test scenario for cinder as a storage destination","message":"Add test scenario for cinder as a storage destination\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/901187\nChange-Id: Icdae86c7aa6138a5541a335f929ab9a90723b194\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/19b4f673b7e832f4c9c18cf22b9c74b02330cf44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/19b4f673b7e832f4c9c18cf22b9c74b02330cf44"}]},"branch":"refs/heads/master","description":"Rebase"},"6fae2b87ad736a2dd26f4ee387918fa0f55e97c9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-29 11:05:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/901189/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/901189/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/4"}}},"commit":{"parents":[{"commit":"5c47790946b26e1f037bd976af0a5f2725fa176b","subject":"Define lock directory for oslo_concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5c47790946b26e1f037bd976af0a5f2725fa176b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-29 11:05:54.000000000","tz":0},"subject":"Add test scenario for cinder as a storage destination","message":"Add test scenario for cinder as a storage destination\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/901187\nChange-Id: Icdae86c7aa6138a5541a335f929ab9a90723b194\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6fae2b87ad736a2dd26f4ee387918fa0f55e97c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6fae2b87ad736a2dd26f4ee387918fa0f55e97c9"}]},"branch":"refs/heads/master","description":"Rebase"},"583dc1390deb8638f423a929399d64b02cc3b057":{"kind":"REWORK","_number":5,"created":"2024-04-03 11:54:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/901189/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/901189/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/5"}}},"commit":{"parents":[{"commit":"5c47790946b26e1f037bd976af0a5f2725fa176b","subject":"Define lock directory for oslo_concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5c47790946b26e1f037bd976af0a5f2725fa176b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-04-03 11:54:18.000000000","tz":0},"subject":"Add test scenario for cinder as a storage destination","message":"Add test scenario for cinder as a storage destination\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/901187\nChange-Id: Icdae86c7aa6138a5541a335f929ab9a90723b194\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/583dc1390deb8638f423a929399d64b02cc3b057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/583dc1390deb8638f423a929399d64b02cc3b057"}]},"branch":"refs/heads/master"},"6af32a409b2519f908c7a68927d4c8b28eadefeb":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-03 12:05:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/901189/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/901189/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/6"}}},"commit":{"parents":[{"commit":"dc5845335d73d64c0a32a0b11a263b027ed1f300","subject":"Define lock directory for oslo_concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/dc5845335d73d64c0a32a0b11a263b027ed1f300"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-04-03 12:05:59.000000000","tz":0},"subject":"Add test scenario for cinder as a storage destination","message":"Add test scenario for cinder as a storage destination\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/901187\nChange-Id: Icdae86c7aa6138a5541a335f929ab9a90723b194\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6af32a409b2519f908c7a68927d4c8b28eadefeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6af32a409b2519f908c7a68927d4c8b28eadefeb"}]},"branch":"refs/heads/master","description":"Rebase"},"f9b03dcb7681c1175e276052abf6b18af8aa4749":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-05-20 10:49:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/901189/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/89/901189/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/89/901189/7"}}},"commit":{"parents":[{"commit":"35aa82224627d0245cab0f5bb51feb0f9f48b825","subject":"Add tag to enable targeting of post-install config elements only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/35aa82224627d0245cab0f5bb51feb0f9f48b825"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-16 17:12:15.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-20 10:49:36.000000000","tz":0},"subject":"Add test scenario for cinder as a storage destination","message":"Add test scenario for cinder as a storage destination\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/901187\nChange-Id: Icdae86c7aa6138a5541a335f929ab9a90723b194\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f9b03dcb7681c1175e276052abf6b18af8aa4749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f9b03dcb7681c1175e276052abf6b18af8aa4749"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
