)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~901562","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2F2023.1~I4ee3fc33fdbeb50fc7b102bf62d6134f83c5925f","project":"openstack/openstack-ansible-os_glance","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-22 17:40:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ee3fc33fdbeb50fc7b102bf62d6134f83c5925f","subject":"Add glance_bin to rootwrap defenition","status":"MERGED","created":"2023-11-21 13:36:26.000000000","updated":"2023-11-22 17:41:18.000000000","submitted":"2023-11-22 17:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901562","cherry_pick_of_change":900930,"cherry_pick_of_patch_set":1,"meta_rev_id":"4a50a0fe40c8731280ac59b4a61c5ccf8e485294","_number":901562,"virtual_id_number":901562,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-22 17:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-11-22 11:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-22 14:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-22 14:57:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 16:13:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-22 14:57:23.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"29103bf4c5b8d4a1daa8bb5140078ced984da111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-21 13:36:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ead9f1c06f790b122b46a6fac7cf8a53da98a97a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 16:13:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94270e48ecf742f685dcee51d5ee459e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2550adf9d5704fcc9517a2a579dfee88 : SUCCESS in 13m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7258a95200ec4e31952811850ae6c0e8 : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6054756decd4dda9bb79812021c32f8 : SUCCESS in 1h 29m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c2647dd6f6ae4d1a91f043ac0434e435 : SUCCESS in 1h 42m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73f155ea24654a60a6d7391e7d7096c3 : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b29bd199fe14dfcb7a2cffef2a87a33 : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/670801c16eee4f88babd64a064bf45ad : SUCCESS in 1h 40m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af42b29f4e28468d94a3dbd094e2fbd2 : SUCCESS in 1h 39m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5ba3edafcb64780b0b30afc2d2ea11a : SUCCESS in 2h 10m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90d30a6c3ddc4108ab72038be13ee2c7 : SUCCESS in 2h 07m 17s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66e7928ff4414fb798e40b20dd696709 : SUCCESS in 2h 05m 38s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb522d6596354c599516ff877360d41f : FAILURE in 1h 51m 42s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b112b03e21a342eb827889ba6a89e4f4 : SUCCESS in 1h 55m 43s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f87666a3d4f04b419422af343b9f8b79 : SUCCESS in 2h 18m 52s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45215e6e5f634dd88bdb74400c642bf8 : SUCCESS in 2h 26m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83db317ff8dc42888e83cf64eece0962 : SUCCESS in 1h 02m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4080bd3323d472eac8c62dea55390cb : SUCCESS in 56m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ec071ba8044308b2cfe8c61cc5d136 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e3d767b166c176aa61b29e5615c3fa07f899c9f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-22 11:54:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4fdbe4a06c448852eac81f402f8a6477874df05","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-22 11:54:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc9cf2f4b4e99d00908fe303c8723cf6e6c50796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:26:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1fc7915f3b545f09707555fafb9779e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44b0c9208c42452da3f16138f55e45d2 : SUCCESS in 11m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/600db1aa12cd4ab8bd1809f95d210b64 : SUCCESS in 1h 10m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e8a7186aa12430ea70a00fc12ca27c9 : SUCCESS in 1h 19m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5bbe9b74db254f9cbc9b6d78d36fbd4e : SUCCESS in 1h 34m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40d15b0a3dde438395dad7040a6910b5 : SUCCESS in 1h 47m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0b59f2218f94454a6384eeba3357a3f : SUCCESS in 1h 29m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7db7ddcdc8254f42973e3b3a0b80d3b9 : SUCCESS in 1h 34m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79722dc5b5be4b5fac62afe2fbf0d4f0 : SUCCESS in 1h 29m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b45625656a4b464b9a6ab802564a9dfd : SUCCESS in 2h 17m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b7130590a504e4689e78e7058cba1c7 : SUCCESS in 2h 08m 52s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d277fd7b7d64fd99d7a1d5ea3603319 : SUCCESS in 2h 04m 26s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a92d97e481c455286707515939b5d2e : SUCCESS in 2h 14m 24s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11c21c44008b40a5b3948ee8b7af662a : SUCCESS in 1h 58m 00s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1e277f6a6c84abaad3c89ab51a93cbd : SUCCESS in 2h 22m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/127bf20e5bc4469384c123dc26aac5ea : SUCCESS in 2h 10m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ebb1c1e0500406f929cf787c5cabbcd : SUCCESS in 1h 06m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b081540c61943a78ec3e0a2d47b9300 : SUCCESS in 55m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9592331b8eef419eaab69453a79ffbbd : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7723b2e474f85f4fa5df96199a2e65ef92265b5a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-11-22 14:57:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9fa4904882f74090f9c559fc512694fdb40c632b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:59:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c410c77837108103503d2f99f8c8c7b01e8ba48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 17:40:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/514c1319be024f8c9d4e42a2b1e8220b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e24eb10e64c74f18b99100f07596cdad : SUCCESS in 11m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3aa26dad213840c8b867cd744b7d23cf : SUCCESS in 1h 38m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/655a8ca196bb40219163a6f59930d189 : SUCCESS in 1h 41m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cbc3b5109544f90b52e0745a0ebaf9a : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0d9bf031ed8d484aa3df9a2d7f7e1d94 : SUCCESS in 1h 39m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8752dff7634741649924716c2385d69d : SUCCESS in 1h 46m 05s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e5586cf686047ceb834bd1e2118326b : SUCCESS in 1h 57m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93669b6a54764652a4b05ad9bb386391 : SUCCESS in 2h 33m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3158b8a4f784f3b9b5f105495756efc : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"aa5bcbdf3b8557e8f6a7a014c49ea4da4a7f0ac9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 17:40:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a50a0fe40c8731280ac59b4a61c5ccf8e485294","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 17:41:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b976cd27cae142d0addbb9999778a5fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b6424267124f4aa3acbc37cde76bef : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5677359abd57731d1004286e520c10b7c86effc","revisions":{"a5677359abd57731d1004286e520c10b7c86effc":{"kind":"REWORK","_number":1,"created":"2023-11-21 13:36:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/901562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/62/901562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/901562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/901562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/901562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/62/901562/1"}}},"commit":{"parents":[{"commit":"f0b9c42a246928325ca7bce9c82c8c3735c8f37b","subject":"Use correct default for glance_mount_points","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f0b9c42a246928325ca7bce9c82c8c3735c8f37b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-14 17:49:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-21 13:36:26.000000000","tz":0},"subject":"Add glance_bin to rootwrap defenition","message":"Add glance_bin to rootwrap defenition\n\nAt the moment rootwrap can not find privsep-helper binary as it\u0027s located\nin glance bin directory, which is inside the virtual environemnt.\nIn order to properly use privsep we must define venv bin directory in\nallowed exec_dirs of rootwrap.\n\nThis also introduces new variable  `glance_rootwrap_conf_overrides`\nthat allows to manage some extra overrides for rootwrap if needed.\n\nCloses-Bug: #2043503\nChange-Id: I4ee3fc33fdbeb50fc7b102bf62d6134f83c5925f\n(cherry picked from commit c2428ab8da9cc3868b5ae86140a63e4a33e28eca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a5677359abd57731d1004286e520c10b7c86effc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a5677359abd57731d1004286e520c10b7c86effc"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
