)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~901563","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2Fzed~I4ee3fc33fdbeb50fc7b102bf62d6134f83c5925f","project":"openstack/openstack-ansible-os_glance","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-23 11:05:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ee3fc33fdbeb50fc7b102bf62d6134f83c5925f","subject":"Add glance_bin to rootwrap defenition","status":"MERGED","created":"2023-11-21 13:36:48.000000000","updated":"2023-11-23 11:06:40.000000000","submitted":"2023-11-23 11:05:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901563","cherry_pick_of_change":900930,"cherry_pick_of_patch_set":1,"meta_rev_id":"5bb7c566698121ccbb7bd00db4eacbc24f036eed","_number":901563,"virtual_id_number":901563,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-23 11:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-11-23 08:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-11-22 11:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-11-23 08:41:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 15:59:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-23 08:41:18.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"45ad936cf8fb967665a0a86142bc65a8f55f52d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-21 13:36:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ad23a137b3675df7d6acc5eda4f55303e9977822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 15:59:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60578f9be5d848e7a3e1ffac1597fba7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f768db2093624d65a934c02064a18c45 : SUCCESS in 10m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f2fb017ebcfe4cef8235d55581a72007 : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7466eaa41c5c493d8a02f741431a4127 : SUCCESS in 1h 24m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df2a1b279142451f96c1d39a8d84b1d2 : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfe04f768335493eaf42ebfd7a1f3bd9 : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c8dc08938f04fe5be181dce48134bf0 : SUCCESS in 1h 22m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98b0d0ab0c3f4f9b826fddbd613a3aac : SUCCESS in 2h 15m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5154aa72a18a42c8a33147aae0eb1bbc : SUCCESS in 1h 42m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bafa92a6c4d94478bc4297f6052b2e52 : SUCCESS in 2h 05m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/902f96abf391470ea273522fcfcdeeb5 : SUCCESS in 1h 47m 30s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad73982fb218409288ba48f1b447f4b9 : SUCCESS in 2h 11m 39s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb4b18fe2bc84a80a92e61ec2c6a88ae : SUCCESS in 2h 11m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5919cfb8c9814fb9adf3362e54b8bdac : FAILURE in 43m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/857f240e69cb44e298ec6a51832cc552 : SUCCESS in 1h 00m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31922ad28faf4c60871fff30afc9f566 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"86d59260bcc099452b2ea4d4f61a636e22cf3181","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-22 11:54:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56d9b4c28a3adcf2cc2128091e5e13cae902679e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-11-23 08:41:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4279bacbf29577dbd9a8d3fcf48b3f6e80ca66c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 08:43:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5e9a27b7ff1e093ac0c7d89022d9756ddf2e939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 11:05:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd28846d4bc4723b8d82fd5cbde8d5c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f6e24dc177d45fa9b467cec74c42fe1 : SUCCESS in 9m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e4d541ba2b294c638b1a661ac9c4ed05 : SUCCESS in 1h 06m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6398e13690734086921810da6d96da78 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/098cadbcf25744d6ba5f71cf8aedb391 : SUCCESS in 1h 01m 49s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e969f195ed47489da9a483f9161c23bb : SUCCESS in 1h 48m 16s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73058a55f95744e08536618fb6b7ffeb : SUCCESS in 2h 16m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e14720bc484f53a716e009febce9c4 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c7e3db5b516e50725668a821239418be11228724","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 11:05:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5bb7c566698121ccbb7bd00db4eacbc24f036eed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 11:06:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f5f839869c40b1b2f0e8859e5a60a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fbaa14d4524cf0b28eb2e04fb80210 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e86ce26c50f365b0a4a857ec42fe5aa444860f9b","revisions":{"e86ce26c50f365b0a4a857ec42fe5aa444860f9b":{"kind":"REWORK","_number":1,"created":"2023-11-21 13:36:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/901563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/63/901563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/63/901563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/63/901563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/63/901563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/63/901563/1"}}},"commit":{"parents":[{"commit":"d6e25e9e457e767fea063ec80145cc4922370282","subject":"Use correct default for glance_mount_points","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d6e25e9e457e767fea063ec80145cc4922370282"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-14 17:49:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-21 13:36:48.000000000","tz":0},"subject":"Add glance_bin to rootwrap defenition","message":"Add glance_bin to rootwrap defenition\n\nAt the moment rootwrap can not find privsep-helper binary as it\u0027s located\nin glance bin directory, which is inside the virtual environemnt.\nIn order to properly use privsep we must define venv bin directory in\nallowed exec_dirs of rootwrap.\n\nThis also introduces new variable  `glance_rootwrap_conf_overrides`\nthat allows to manage some extra overrides for rootwrap if needed.\n\nCloses-Bug: #2043503\nChange-Id: I4ee3fc33fdbeb50fc7b102bf62d6134f83c5925f\n(cherry picked from commit c2428ab8da9cc3868b5ae86140a63e4a33e28eca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e86ce26c50f365b0a4a857ec42fe5aa444860f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e86ce26c50f365b0a4a857ec42fe5aa444860f9b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
