)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~906047","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ibcd9df6a547febb8f47b88d0c98277b46faf489c","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/glance_uwgi","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-01-27 01:02:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibcd9df6a547febb8f47b88d0c98277b46faf489c","subject":"Enable uWSGI usage for Ceph backends","status":"MERGED","created":"2024-01-18 14:27:25.000000000","updated":"2024-01-27 01:03:14.000000000","submitted":"2024-01-27 01:02:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906047-osa/glance_uwgi","meta_rev_id":"5049b54485ff5cd19b360ae615ad479e9b361eae","_number":906047,"virtual_id_number":906047,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-27 01:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-01-19 09:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-26 23:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-26 23:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-18 16:57:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-19 09:44:27.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-01-26 23:00:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"4f625a66fcd6bfa28019ee513058d7ae74a667e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-18 14:27:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0db3e81fb9a375c428106ce304fa3260535ab85e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-18 14:27:42.000000000","message":"Topic set to osa/glance_uwgi","accounts_in_message":[],"_revision_number":1},{"id":"3eca627b41ab35bd26c1332f3dd802a40c10acc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:57:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49993f23c9c945d7a5b0e8ef3ba37d17\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b18437a4b2548f8bd69eefd0c1eecc4 : SUCCESS in 18m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b604bd70ac894aa19868a6f45ba2381c : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/566ad21bf7254fb191a0c8310ecdc8fb : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b67eea7c2fb4a92bd0f1bf99ac372ed : SUCCESS in 1h 39m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b3aa469bc944f85b4cb374ee227fa3b : SUCCESS in 1h 34m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/664e959290a040c38ea59d102bab22ba : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c6f6efc9cccd47e2b47dfec623ae98ba : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df3ceb14791d46809f2203ff92851a29 : FAILURE in 1h 01m 53s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ea04344fe4c4131b97b757867a6ec84 : SUCCESS in 1h 21m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/deea3bccbb36479fa2abeb0c22614caf : SUCCESS in 1h 59m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39da4b420b2d4713a539bb29e2ad4783 : SUCCESS in 1h 50m 43s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f0c78c9968c44fc801b411eddf78f3e : SUCCESS in 2h 02m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c94e05af34f948e991437eba21db5646 : SUCCESS in 2h 06m 28s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca439ac39afa4f4eb8d28c91eec5eec5 : SUCCESS in 2h 09m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32ad372a11a2448aac217d1f8504d39b : SUCCESS in 2h 05m 43s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a284438762e4635984490646785b94b : SUCCESS in 1h 00m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f1abef811664886bbbec0d505caa829 : SUCCESS in 57m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f52ccf1638149b89e4c5ec692b1977b : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb12c5b3f44b42d0a5ef7779c5cf7612 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9cf4385be901f583810bc5c5e01c716013db0b5e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-18 20:04:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6247427a2f4a5d19d1eb423316ce54645ad0875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 22:41:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b598baf1ad5241fca2acec7556c8cc88\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdfa22442ccd4913a60af48784111332 : SUCCESS in 11m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fd7c6c6077d045f3b780a45184559bb5 : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/12b271f72e584ee082abab30f7f8b9f4 : SUCCESS in 1h 11m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/afaee772be904af5a653a2ffffd5f65c : SUCCESS in 1h 24m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf24cda015254a4f96f8f67beb21e505 : SUCCESS in 1h 28m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2282b95084e41a38d58a53e6f84e5e6 : SUCCESS in 1h 21m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/336545bd273643d997b229e19b7cba23 : SUCCESS in 1h 27m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d68857e4e3d94077b4fc202f8246795a : SUCCESS in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/932679b8eaea49d699c6290c8a855b95 : SUCCESS in 1h 24m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf0cc3248c104805a370e963f179c252 : SUCCESS in 1h 59m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b064910f3a824532aea5edcbaec52296 : SUCCESS in 1h 53m 38s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6810858be26c44a69c6e03e184504dc6 : SUCCESS in 2h 06m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad8188984cc34118be69c895a58bf448 : SUCCESS in 1h 28m 49s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a1955a9d41d466c83f9ac67d3fd7c0c : SUCCESS in 1h 59m 52s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/292ebf14882d45be821c17fd8795a718 : SUCCESS in 2h 15m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/022027abf7b4421a9fedda3ef04b80a0 : SUCCESS in 59m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f690b69c9eca49449f71afa8c47b3d9b : SUCCESS in 48m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342a8acf191c4e249347626938ef9967 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9520efd0e65463aad4c8ad6d9afd78b : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7e5960aec0e7850a7e1596c0e6a76a0ded744453","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-01-19 09:44:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d975930fc471606f2c67318d8687f72ed0bd19c4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-01-26 23:00:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dded2636570cc6ee0ee63dc8617d4d5ac04f02d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 23:02:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ad0a6a8a5b3b2f890d953560aaeb4f77811a7f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 01:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086e9e3b19f048dfb019e5ef4a2f9dda\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/750eb978f6a74171980d38af1ccaac4c : SUCCESS in 12m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d15b2414485b421aa1dc5a4f9d771f94 : SUCCESS in 1h 14m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/672ad3b16bbf4d1183a98a963b2e8def : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47500b1c6e254fa295c2faa35cb9d5ef : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9fb5283008d4855b4947d3a2b71230d : SUCCESS in 1h 15m 10s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c25b714e2e0e45aaa7b3340021462b84 : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bdf42b3ca3724949ad4223b166a14bba : SUCCESS in 1h 23m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e1127dbb10745f7b802a19e60cb8516 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ee1b2f45e3f4f6ab4bf934784555b63 : SUCCESS in 1h 58m 12s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38fad83857c246ceb09e9d36dd709364 : SUCCESS in 1h 42m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a948cddf784a25bd4a0b007b03f131 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bd4e8b774b241b398c477cbe577af80 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4f6d83bb744b6980dfe21ae2fb641aaa6ce0ff63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 01:02:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5049b54485ff5cd19b360ae615ad479e9b361eae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 01:03:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779ad096a27f4d43b63bac271d165f47\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095c82599ac04ebb8e10295a7f17818b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0f6fd67cc732f0847b027a2bb136fc8dcde72b1","revisions":{"d0f6fd67cc732f0847b027a2bb136fc8dcde72b1":{"kind":"REWORK","_number":1,"created":"2024-01-18 14:27:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/906047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/47/906047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/47/906047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/47/906047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/47/906047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/47/906047/1"}}},"commit":{"parents":[{"commit":"e4acad2bb660d2776bc86e30e7791f9f9adb5c4e","subject":"Fix cinder store configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e4acad2bb660d2776bc86e30e7791f9f9adb5c4e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-18 14:27:17.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-18 14:27:17.000000000","tz":60},"subject":"Enable uWSGI usage for Ceph backends","message":"Enable uWSGI usage for Ceph backends\n\nSince upstream bug [1] preventing us to disable uWSGI for Ceph was fixed,\nwe can remove extra logic of disabling uWSGI usage when Ceph is among\nconfigured storages.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/885581\n\nChange-Id: Ibcd9df6a547febb8f47b88d0c98277b46faf489c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d0f6fd67cc732f0847b027a2bb136fc8dcde72b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/d0f6fd67cc732f0847b027a2bb136fc8dcde72b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
