)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~909820","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I9b4c0985969a4945ce8a1058206376b0347fbb27","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"property-protections","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-02-28 09:31:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"35983":{"account":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"last_update":"2024-03-06 15:18:01.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-03-06 13:21:29.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I9b4c0985969a4945ce8a1058206376b0347fbb27","subject":"Add property protection configuration","status":"MERGED","created":"2024-02-22 07:15:24.000000000","updated":"2024-03-06 15:19:04.000000000","submitted":"2024-03-06 15:18:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"909820-property-protections","meta_rev_id":"f6f12fc5495cd2b64da24192b8e23bbeb4bafbe8","_number":909820,"virtual_id_number":909820,"owner":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 15:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-02-28 09:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-06 13:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-06 13:21:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 10:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 16:08:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2024-02-24 09:18:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-02-28 08:07:30.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-03-06 13:21:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"a7f11dc1eec54a106afc699a327f409a7865b77b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"date":"2024-02-22 07:15:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faa07625816672b3b627474d5f68873a33dd998d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 10:23:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d083da1683a140a19f35aab726b4721d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd4a1f19eb3e4b2f819aeeb22cbed784 : SUCCESS in 14m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ff8d0b19cce64f808754d8c4c1ecb82b : SUCCESS in 1h 12m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/13ae0455e5414778944db78cb21dc8eb : SUCCESS in 1h 19m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0ff873f2b9c43d5bd37311ec8c90f4b : SUCCESS in 1h 25m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7b9fff37470a4efea533aa9b840db259 : SUCCESS in 1h 35m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/caa937b632f9427ca7abe9760ede8e85 : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c9473578bf64d2280cf4ff5a0bac58a : SUCCESS in 1h 19m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b65e456105243d99b761b3a4c341de9 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b97beb357405480b8d4acf6b003e0bfe : SUCCESS in 1h 16m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6441bed536b645c3b49856ab72b77881 : SUCCESS in 2h 09m 47s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f145742d3bc4b4e83abfb331a5e7fa7 : SUCCESS in 2h 59m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be24f228e52c42baa666ff41e181ac9b : SUCCESS in 1h 46m 52s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bdf2132a89f439380eec8ce8611b182 : SUCCESS in 2h 21m 56s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63be7e0d8109424eb03170127b69f96e : SUCCESS in 1h 57m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/526220382f48478ea947b553baa16f16 : SUCCESS in 1h 49m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b716f61202b044d99f64e6f04cb2e29f : SUCCESS in 1h 54m 32s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/224784850d264172a2d0b1279938e53a : SUCCESS in 2h 13m 39s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39e07edbba5f48e3bce275a139aeb1cb : SUCCESS in 1h 01m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d39c64ef313f462389cf975600941d22 : SUCCESS in 58m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432323cee7c442e4842d8b0670500d80 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6c9fc1fcf89a949da96c18adfff13bd514565337","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-22 16:08:59.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"af669180cf56314b6a1f5b927ceb68a2c6a14b50","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-02-24 09:18:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78b6bb6a6125bdf6b67128efc17f8f596c7bd68b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-02-24 09:21:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f97aadac7434c54feb24891def7a78ee95e1de00","author":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"date":"2024-02-26 08:12:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"34a8316b08115de2d9d488b1651e7182d7a68607","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-27 12:15:55.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d9a36a7eae8b5bcf70506b40866a99ae1d20beda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"date":"2024-02-28 07:09:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"68f3b3c550cca7e2ba6c9062cbbf9b1d5d84d910","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-28 08:07:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a6955e506053446f3047c85214b3252518532273","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-28 08:08:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"31e6b50f806bb5a62d2f543ab885917bf80e6521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"date":"2024-02-28 09:31:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"98a80fd122d7da6741ca61729ca89b7f87fc7faa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-28 09:31:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce42e4a26bc893ad3adf22099fcd656c97cccb02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:12:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a7ab7fcad384d7b9ba645c4944d5ad0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca855c0e787041e1b673cb7d127cbdc4 : SUCCESS in 14m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3199f8b9599c464ba38014832f036617 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f58c4b70805475abd6c92f2c33c7fe1 : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/716cf093af714c79bb5b9d33b9a16d1d : SUCCESS in 1h 32m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d930ca061e7345baa4cce5e16c680832 : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d4f02c0fc6544209e04feee7c759d1c : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e406953393245e3880a2eee03fed2fe : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cc28f2b173b4cafa7a75bc9ec0d4f81 : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe63c98ae2af45849bbe409be6190121 : SUCCESS in 1h 17m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55a6e2e694a043bab7b8f567877532d5 : SUCCESS in 2h 13m 48s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/506243192e1d4740af94e49bd0a59dbc : SUCCESS in 2h 36m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48b92e0f7d234b6fb936e553b0e56d6e : SUCCESS in 1h 52m 41s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b490251c35b457bbc7c5986444ed0c4 : FAILURE in 1h 20m 48s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8fe89819dc064d38a458483c4517ec44 : SUCCESS in 2h 25m 09s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f888a44cd0e46ef886a9899632b822b : SUCCESS in 1h 56m 01s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9a4ea560264d579c53249202a4892c : SUCCESS in 2h 20m 35s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a1cbe4b525a45c49c70eb3ff0716b46 : SUCCESS in 2h 05m 04s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd5d3e8a579740e88cb064f5c0bc2cf4 : SUCCESS in 1h 09m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2eec9ee15c1413394ac52773e7ca9e8 : SUCCESS in 1h 05m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8af91e035df4f39bae3732a409d20ec : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4369050914ea4542b809b17581979ee5 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"82e3fbaa4c79458446c762ebdb1c5b07db3ff0cc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-06 09:34:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a850f47d2834cdf9aeca6b6f1da2b15b9f7224d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:15:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e1f1aa5272e4d0d80d193a210d8bce7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83faf3ff5cb24e5487afa5ce28c0de28 : SUCCESS in 11m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/66073b1b757243b291cf05d39cc7157a : SUCCESS in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/58e8b220b2714138b2a9efbdabcd9940 : SUCCESS in 1h 13m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccfbb30424e84e3d9c30a98f8eaa210c : SUCCESS in 1h 32m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd5959b9494b4b74b1c4f00bd672b615 : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7e57bfbedad484db17500e4bde6b171 : SUCCESS in 1h 08m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9094fbda211d460c8f51ada3df56530c : SUCCESS in 1h 37m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e696305d3c0f4e3db83cdb8e40c68aee : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15b938e44b11491aaf2d075e9cbc7727 : SUCCESS in 1h 13m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3f02033ec5746f2979dcc4cbdd0b97b : SUCCESS in 1h 53m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a697da68f3c04899bfa4124df45ec625 : SUCCESS in 2h 36m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9bea9989e814231923b0b5ec2e13f3f : SUCCESS in 1h 46m 52s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/030317aa77f942ffb2bf7beb63731f54 : SUCCESS in 2h 11m 58s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fe060ba02a4c439b9011fec9cda1ddc7 : SUCCESS in 2h 03m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce4cf1a17e9846c0b60f5c91cb1908cf : SUCCESS in 1h 38m 00s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0f67b125f6c4694a37814e52fef80c8 : SUCCESS in 2h 06m 55s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/285a6016cf974edc829d52ce0da810cb : SUCCESS in 2h 15m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa5c2d6641584476ba895045e8b102c5 : SUCCESS in 1h 01m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaf3f3447dc9439e9531fdf662c1ff16 : SUCCESS in 1h 03m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6eb29c8481c4ae29386e29f7e0c3106 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/859f6ad8925d4392beba3bbd6bd5d93b : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":3},{"id":"5c52ff2d5be6d7900ffbf0cbd02d80e11a6e3352","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-03-06 13:21:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9531de6b97a5f19d445681df4461a841fed80d05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 13:23:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fdbf97d4a0e4a58de5f28ce237bd44f5ea8b26d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 15:18:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008ec875efc6420d99b0799613eb765b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/640175aacfa547f7a163069856dc0ada : SUCCESS in 11m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e5cba29691474c23b3a3069d2bb4f342 : SUCCESS in 1h 23m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6161feabb9af444cb7a209683d95e141 : SUCCESS in 1h 05m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8dd4c330bef4c85a490f29d395b3eea : SUCCESS in 1h 29m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24305e8d27df4dc19f53c0908e409fdd : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9033d8497e241fd90726ba574165667 : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/95fe11c980c0426cb5e459b35dce251a : SUCCESS in 1h 21m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efcdba886acd46759aa4550f2134b2d4 : SUCCESS in 1h 15m 16s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5916c04af13c444694850ab205043c75 : SUCCESS in 1h 48m 56s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe333e1d04264484beab629451756493 : SUCCESS in 1h 53m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a15ba7ed31421d907e787eb50725de : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ae4aa5e833e4303940146fb21104368 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"7ed57e229ce68139cb3cb87695f5cbca2ece129f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 15:18:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f6f12fc5495cd2b64da24192b8e23bbeb4bafbe8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 15:19:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aba2f33bc4fd4a1c9a6b00d1fe40e659\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cafa614804a54a62840df2e36c1e84b5 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e8de2a5f0b2faeb0774e60d5c5c8ff15b1139c02","revisions":{"3fbbfc81b7774717d424c9dcfcc58b90ef129086":{"kind":"REWORK","_number":1,"created":"2024-02-22 07:15:24.000000000","uploader":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"ref":"refs/changes/20/909820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/20/909820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/1"}}},"commit":{"parents":[{"commit":"f375f499c03d8df2b088d8d76670dd3164ed24d8","subject":"Merge \"Enable uWSGI usage for Ceph backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f375f499c03d8df2b088d8d76670dd3164ed24d8"}]}],"author":{"name":"Aleksandr Chudinov","email":"silentirk@me.com","date":"2024-02-22 07:14:15.000000000","tz":120},"committer":{"name":"Aleksandr Chudinov","email":"silentirk@me.com","date":"2024-02-22 07:14:15.000000000","tz":120},"subject":"Add property protection configuration","message":"Add property protection configuration\n\nAllows to configure property protection according to\nhttps://docs.openstack.org/glance/latest/admin/property-protections.html\n\nChange-Id: I9b4c0985969a4945ce8a1058206376b0347fbb27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3fbbfc81b7774717d424c9dcfcc58b90ef129086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3fbbfc81b7774717d424c9dcfcc58b90ef129086"}]},"branch":"refs/heads/master"},"ff5420acef132865410def44bc2698a7c40d29ca":{"kind":"REWORK","_number":2,"created":"2024-02-28 07:09:01.000000000","uploader":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"ref":"refs/changes/20/909820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/20/909820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/2"}}},"commit":{"parents":[{"commit":"f375f499c03d8df2b088d8d76670dd3164ed24d8","subject":"Merge \"Enable uWSGI usage for Ceph backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f375f499c03d8df2b088d8d76670dd3164ed24d8"}]}],"author":{"name":"Aleksandr Chudinov","email":"silentirk@me.com","date":"2024-02-22 07:14:15.000000000","tz":120},"committer":{"name":"Aleksandr Chudinov","email":"silentirk@me.com","date":"2024-02-28 07:08:42.000000000","tz":120},"subject":"Add property protection configuration","message":"Add property protection configuration\n\nAllows to configure property protection according to\nhttps://docs.openstack.org/glance/latest/admin/property-protections.html\n\nChange-Id: I9b4c0985969a4945ce8a1058206376b0347fbb27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ff5420acef132865410def44bc2698a7c40d29ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ff5420acef132865410def44bc2698a7c40d29ca"}]},"branch":"refs/heads/master"},"e8de2a5f0b2faeb0774e60d5c5c8ff15b1139c02":{"kind":"REWORK","_number":3,"created":"2024-02-28 09:31:11.000000000","uploader":{"_account_id":35983,"name":"Aleksandr Chudinov","email":"silentirk@me.com","username":"silentirk"},"ref":"refs/changes/20/909820/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/20/909820/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/20/909820/3"}}},"commit":{"parents":[{"commit":"f375f499c03d8df2b088d8d76670dd3164ed24d8","subject":"Merge \"Enable uWSGI usage for Ceph backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f375f499c03d8df2b088d8d76670dd3164ed24d8"}]}],"author":{"name":"Aleksandr Chudinov","email":"silentirk@me.com","date":"2024-02-22 07:14:15.000000000","tz":120},"committer":{"name":"Aleksandr Chudinov","email":"silentirk@me.com","date":"2024-02-28 09:30:50.000000000","tz":120},"subject":"Add property protection configuration","message":"Add property protection configuration\n\nAllows to configure property protection according to\nhttps://docs.openstack.org/glance/latest/admin/property-protections.html\n\nChange-Id: I9b4c0985969a4945ce8a1058206376b0347fbb27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e8de2a5f0b2faeb0774e60d5c5c8ff15b1139c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e8de2a5f0b2faeb0774e60d5c5c8ff15b1139c02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
