)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~914752","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ibe3eb5d50f8e0904d73b93713dda2be0a6b38df7","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-04-25 16:42:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe3eb5d50f8e0904d73b93713dda2be0a6b38df7","subject":"Define lock directory for oslo_concurrency","status":"MERGED","created":"2024-03-29 11:03:45.000000000","updated":"2024-04-25 16:43:49.000000000","submitted":"2024-04-25 16:42:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914752","meta_rev_id":"8d763053684f4953392bd2523abe9b0074768bbc","_number":914752,"virtual_id_number":914752,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 16:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-04-25 14:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-25 14:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-25 14:10:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-04-02 14:37:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-29 13:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-25 14:03:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-04-25 14:10:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"d9e1b3719d007df7b957a84d3c230f52ed617c90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 11:03:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"475001d65095a180fba39f200918b4018d216ccc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 11:03:53.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"ec0224918750954a407a58104068df08bd51b9ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 13:35:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87d5deebe420485eb5bd508654c5410e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3b817ece43d4d4cb7af508293a40403 : SUCCESS in 13m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c1c907962ca24c06846586253d242914 : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ef2e58ed733c4cf791ee05d0471f235e : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/70d8adcd967441188026d00fffb13813 : FAILURE in 1h 19m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9f97d9582364fe09cd1a98ca89fa572 : FAILURE in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f7579704e0040c1bc31d20272b9d74c : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3631e72c713a4ef689b4199e5e3f077d : FAILURE in 58m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66c6b149e2f64ba1987d26a3049f1f02 : SUCCESS in 1h 14m 56s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80624667725e4fe088791e5cf7b347e1 : SUCCESS in 1h 12m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28bedfc7ef9949ceaeae27b9769cb75d : FAILURE in 1h 41m 36s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/057c1ec086244c228ad61146deebed60 : FAILURE in 2h 14m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfe1e803789647f4832299fbe95b52d9 : SUCCESS in 1h 31m 23s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c40fb9e8f8d4c46a35605fdfb130247 : SUCCESS in 1h 56m 49s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96ec0328f6544193b03fea484e629661 : FAILURE in 1h 45m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cb7140fe7db4a0eae13f904add5835a : SUCCESS in 1h 33m 03s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b235229c2387483a97c644af87a22525 : SUCCESS in 1h 55m 53s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d29de4f6db704aaf97340bad33ea57f3 : SUCCESS in 2h 08m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/36ffe172c4204e4f92e1fa7739bb62b0 : FAILURE in 50m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a07c634bfad4f92b41b30a355f1866d : SUCCESS in 1h 03m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2536db0f0a6e4578ae376dcca60306de : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4003f077de041f3e1da89d5424902876a0472b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 14:36:22.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"719a36bb42621f482562b8d760e233aec588755e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 17:11:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a33d4e8a3b784c1ba4c8caf20505462c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4542e744863e4445a74b8a6f392d3e57 : SUCCESS in 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/83a0da51a8ba417688bc6fba8d9198a1 : SUCCESS in 1h 10m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e6b4b918aee4917b81c51b8922c36d4 : SUCCESS in 1h 27m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc22c7869db24de79bca0a97a02eb23a : FAILURE in 1h 12m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a142bcd535a4ed5b522909e24d8b894 : FAILURE in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22777ae4fd0e41eea1f73a8df6bbcef3 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ccb56abd22b44138952b087615a81a8 : FAILURE in 1h 20m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e66fe0d871a44e89c0bd6637bdfd8bc : SUCCESS in 1h 26m 00s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76ad176e06fd40d8854f6cdc63b1ca67 : SUCCESS in 1h 15m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/601c2c1c4a3c45a1b237b65b0920cf64 : FAILURE in 1h 50m 01s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0191473547cc4b95a81ea351631dab66 : FAILURE in 2h 27m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62a8755f30c44d22a1e063dda6baf3e0 : SUCCESS in 1h 43m 56s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f5797c7b08144448474df2ee995f58f : FAILURE in 1h 45m 35s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74d54ccbdfa44190b924f1ce3fd98a59 : FAILURE in 1h 56m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aec9cf239da3430b8f99d91c5e9e1ad3 : SUCCESS in 1h 43m 05s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4a230d142714222b330cd30c76628c3 : SUCCESS in 2h 01m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1167f311f1d4d049939f071b29f5514 : SUCCESS in 2h 03m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4b0bed075a040e3ac58229f5f575c21 : FAILURE in 1h 04m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d3c368c34af482d986429ea1832994c : SUCCESS in 1h 02m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a3041c509645c0b78c1125c2a61b9c : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"ffa398af528d04bc4af67c20a9c4165924341053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-02 11:28:40.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a22571d29b41ecff1cef517584f73f890cdc8447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 14:29:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fbaeae05bf74444b4784939f082ad67\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e902606827e54968954ba386584491bb : SUCCESS in 14m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8d4fd89134df4c888d59c363a57ebbea : SUCCESS in 1h 02m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f1dd414c4e3e44839fdcb73ff240c018 : SUCCESS in 1h 14m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/11edf713274c4e29991fae454a5e4e39 : SUCCESS in 1h 17m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d1a133c54d84813a0afa70f5d353ded : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b6523aa74b34f93a377fdede1206720 : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/da7f2d0d508e4313998fbd69a752f312 : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b95b87e63bce47dfbe21ca69e4fad702 : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edc48cc86c8949f285f2cab50adaf4a0 : SUCCESS in 1h 39m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/495eea571d1e4bc296fc1c06c3e7a014 : SUCCESS in 2h 01m 07s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96d3ea45825641f2b6e1ed156ab8c330 : SUCCESS in 2h 22m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff436e4028d14b29aaf609a80a95a36f : SUCCESS in 2h 01m 38s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16f54ba76f3c4c9abc0d068c98b6a5cc : SUCCESS in 2h 21m 50s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39b6c2ecd9964772aa319738b055b4e9 : SUCCESS in 2h 00m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b4871fe63ce4c5bb2c17f1a2513db4d : SUCCESS in 1h 57m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f31b0eb747c444d9be12a2ff3b91e66 : SUCCESS in 2h 20m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69b251407b274322960cc91880a8bea9 : FAILURE in 2h 10m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/630e3a7815a1416fb3bf9b80f044ef2d : SUCCESS in 1h 06m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afa220347b4843fe81ebefc4f9aa0b5f : SUCCESS in 58m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662c7a345251420ab2299310e2a20067 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"a6c9c0813aedcf9ad8553c99793900bcde0acba4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-02 14:37:15.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22293f50439208174844f187a5ebd00c4480a63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 17:18:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ade32312683422f992cf6a9e560f556\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffb891a794444f8d962ec1167653d407 : SUCCESS in 14m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/55bfc3f7cd1542ad9e5cd36a929e130c : SUCCESS in 1h 08m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ecaed5134f76405fa1c0f0ccd19f9a10 : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4686a0ecace40b8b413a26b579ae021 : SUCCESS in 1h 16m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c16ff281d5ec4e11add24b0a5b7fee9d : SUCCESS in 1h 20m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a731f72cc714f78aa72ad048ef3fb23 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27bd8efebaba4ba49295fedb09c45c48 : SUCCESS in 1h 31m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e80564a7665a418da2b4a0535026b55e : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2beabc1b0d548ed8cfa43fd85ff83b2 : SUCCESS in 1h 14m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/609231f2afe6418cbea42af011fb732a : SUCCESS in 2h 12m 34s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/753a6bd1ae24449e94b1f56f53a6aa0c : SUCCESS in 2h 33m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e4bf91f683e4441b8aaaf6c3cf35519 : SUCCESS in 2h 00m 13s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a8d306102b6436da2301ccdb6adc334 : SUCCESS in 2h 29m 21s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed2cc86d913f44b6a3ffdb4359d4f369 : SUCCESS in 2h 10m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/317ea29dd2844fb092ceef26f4116b5e : SUCCESS in 2h 03m 09s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ac338ecf5ce4a15a3fc0efb5ae1627d : SUCCESS in 2h 08m 03s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1a69570d258467aaee8be10aeb223d0 : SUCCESS in 2h 12m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c32de71001b485e8d1d7a0f85aab54a : SUCCESS in 1h 13m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cc53bfa29d74c648bde2b3061152214 : SUCCESS in 1h 09m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d040372d082443ca373e05ba1ccbb44 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"9a9d31142d8256426c007ec47a7291d31fb01cc3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-04-25 14:03:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2ce5799d9751a7cc7b50901a2b3c8df5d5260c0c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-04-25 14:10:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8fe6a228949f887298dd18268426815d1cdbf804","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 14:12:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e80e634d939730659bed1c197c0d6adfd4c0f1a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:42:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab1795478304fa3b7f83c47aed50e0a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3aa49c559bf4457cad15305c663e3255 : SUCCESS in 14m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/64397158877349d3a96e00466a6114bf : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/503e4088355a4c7385f636d1fdef21cb : SUCCESS in 1h 15m 45s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d1b79756d1574d44b2dc87baf00fac5c : SUCCESS in 1h 33m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e0d02811d5a48e792bee855102e41dc : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eced44934d114fe891134604361659d3 : SUCCESS in 1h 21m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc39eeff4fdf4f288fd2a4104b490f28 : SUCCESS in 1h 33m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb0751fc6ca0476c8d74dde752cf476b : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ddd78c40d474f55a536908fd288227b : SUCCESS in 2h 02m 39s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e6cdd5e88874e9d8095edf74dc04c42 : SUCCESS in 2h 19m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857728f86f4d4d35b02b886bb27234b4 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"fe7b493ce85e2e934896d867709470df9f972841","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:42:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8d763053684f4953392bd2523abe9b0074768bbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:43:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c58da17afe24ed88c343d2e4e3537be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18397b52b35e48a5bcc528e282541c65 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dc5845335d73d64c0a32a0b11a263b027ed1f300","revisions":{"5c47790946b26e1f037bd976af0a5f2725fa176b":{"kind":"REWORK","_number":1,"created":"2024-03-29 11:03:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/914752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/52/914752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/1"}}},"commit":{"parents":[{"commit":"bf25a670ff4a229d7412e85051577f7829168495","subject":"Add worker_self_reference_url to glance configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bf25a670ff4a229d7412e85051577f7829168495"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 11:03:36.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 11:03:36.000000000","tz":60},"subject":"Define lock directory for oslo_concurrency","message":"Define lock directory for oslo_concurrency\n\nWith [1] it become required to define lock directory for oslo_concurrency\nwhich was missing previously.\n\n[1] https://review.opendev.org/c/openstack/os-brick/+/814139\n\nChange-Id: Ibe3eb5d50f8e0904d73b93713dda2be0a6b38df7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5c47790946b26e1f037bd976af0a5f2725fa176b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5c47790946b26e1f037bd976af0a5f2725fa176b"}]},"branch":"refs/heads/master"},"102a2d9166a06dc049ad8c4019f16cb49c793a3b":{"kind":"REWORK","_number":2,"created":"2024-03-29 14:36:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/914752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/52/914752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/2"}}},"commit":{"parents":[{"commit":"bf25a670ff4a229d7412e85051577f7829168495","subject":"Add worker_self_reference_url to glance configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bf25a670ff4a229d7412e85051577f7829168495"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 11:03:36.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-29 14:36:22.000000000","tz":0},"subject":"Define lock directory for oslo_concurrency","message":"Define lock directory for oslo_concurrency\n\nWith [1] it become required to define lock directory for oslo_concurrency\nwhich was missing previously.\n\n[1] https://review.opendev.org/c/openstack/os-brick/+/814139\n\nChange-Id: Ibe3eb5d50f8e0904d73b93713dda2be0a6b38df7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/102a2d9166a06dc049ad8c4019f16cb49c793a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/102a2d9166a06dc049ad8c4019f16cb49c793a3b"}]},"branch":"refs/heads/master"},"dc5845335d73d64c0a32a0b11a263b027ed1f300":{"kind":"REWORK","_number":3,"created":"2024-04-02 11:28:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/914752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/52/914752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/52/914752/3"}}},"commit":{"parents":[{"commit":"bf25a670ff4a229d7412e85051577f7829168495","subject":"Add worker_self_reference_url to glance configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bf25a670ff4a229d7412e85051577f7829168495"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 11:03:36.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-02 11:28:33.000000000","tz":120},"subject":"Define lock directory for oslo_concurrency","message":"Define lock directory for oslo_concurrency\n\nWith [1] it become required to define lock directory for oslo_concurrency\nwhich was missing previously.\n\n[1] https://review.opendev.org/c/openstack/os-brick/+/814139\n\nChange-Id: Ibe3eb5d50f8e0904d73b93713dda2be0a6b38df7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/dc5845335d73d64c0a32a0b11a263b027ed1f300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/dc5845335d73d64c0a32a0b11a263b027ed1f300"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
