)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~916812","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ie427f4c69a7bb3c8460400b0ee88e5d482080786","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-08 16:47:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie427f4c69a7bb3c8460400b0ee88e5d482080786","subject":"Add service policies defenition","status":"MERGED","created":"2024-04-23 16:31:47.000000000","updated":"2024-05-08 16:48:37.000000000","submitted":"2024-05-08 16:47:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916812-osa/messaging_improvements","meta_rev_id":"10720c5c914683d34b3b1127cbdd07e72fb8465c","_number":916812,"virtual_id_number":916812,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-08 16:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-07 13:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-08 07:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-08 07:13:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 19:20:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 13:02:42.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-08 07:13:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e6690094d27adbb4a08d388e5be189d8e85accf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-23 16:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19851a8ffeab46c12d4eb5f80e713df97611e77d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-23 16:32:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42b464213afcae8e8a0c2ab53315713e855dc204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 19:20:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f09e2299ffd40c3981fbc9ae2b57f2e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e614d0714fc428bbf37351af461cd0d : SUCCESS in 12m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/da7e424efda343f6a9b81dc71f7f036d : SUCCESS in 1h 19m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6f641bf94b124006bbfd6a3a76e1cfa5 : SUCCESS in 1h 15m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3efd532086f0444e9e6e36f51ddecdff : SUCCESS in 1h 11m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31184e09688142f1a0c5ff2a8e20125a : SUCCESS in 1h 33m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f13856744854b969ac0a6189eaa4dc3 : SUCCESS in 1h 09m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2aca75f9e0a74c7598e973fbb62e134d : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/154eabdff0aa45cdb5cf2334944d744f : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/204a0c9638474f129831e941b59242ab : SUCCESS in 1h 06m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf2374ffbc6c4341bc69aa176312dc75 : SUCCESS in 2h 08m 20s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/524b7732ab3441ab8924f805817553d8 : SUCCESS in 2h 34m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6582082206f44170b10694a7fe5a5577 : SUCCESS in 1h 57m 26s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06cf4966630e4c8ea5717f3fe0d26fc5 : SUCCESS in 1h 55m 08s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c883f7f186b4e278876208fac5ddd55 : SUCCESS in 2h 11m 10s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63db877e136849259050a4817e62a4cf : SUCCESS in 1h 41m 36s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad856e54bcbf4f78ad010166e9d5954b : FAILURE in 1h 24m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d155e8c94d2b42cab8c9d56000388abe : SUCCESS in 2h 28m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1af4dc0a76534a81b6d0ad6c6dbc2026 : SUCCESS in 1h 10m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/449b9c141399495397fc78fce00f7346 : SUCCESS in 1h 06m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6791293651694f5381685e3e35301969 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"95daa5576c9c876a325f0467cb396d810f792c02","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-24 09:39:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efd5bff427cca37dfd70cd89f74ead9874e8c893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 12:29:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a05c9d63a04a0690f8eb024ac6568e\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aad123dc219c46b4aec6d42f0c8c3af3 : SUCCESS in 12m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/82ac66876af54d2997faf6b33fa53474 : SUCCESS in 1h 14m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a1082d8dc2054c8d92c874e7b26963ba : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/22ad2e4bec0f484c91aedd0b55136520 : SUCCESS in 1h 24m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15293dfda8e145ca910dacf72faf7f1c : SUCCESS in 1h 21m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0b23cffa2bf44a8a844185d1f1e6efe : SUCCESS in 1h 14m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20966312d624457cae84ebff4ce9f086 : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d5a1c39d024b42b3c4379dda873b56 : SUCCESS in 1h 17m 08s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb4b05b7169a4877b2dac3190450fdd6 : SUCCESS in 1h 16m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/743dfe29616d45a7b07e83089435855b : SUCCESS in 2h 05m 11s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c35f2faaa5d748429fe68f8c75a7f6c4 : SUCCESS in 2h 14m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/703c0d5c6a3746cc8408a98558ea4a3d : SUCCESS in 1h 45m 25s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4603eaa218b94fc38c18233192f9ceca : SUCCESS in 2h 22m 56s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4ecf2be9dcb487691120a072058b600 : SUCCESS in 2h 01m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfc88c44e6304ad3994cf55b52637634 : SUCCESS in 1h 28m 29s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9a72a0cf0024ed5a40d9f36daee9b2d : SUCCESS in 1h 53m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9378b3cf0ab647248562c04c35ecb0c4 : SUCCESS in 2h 23m 01s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7149350192cf4f88b4841f3f6f6f6873 : SUCCESS in 1h 07m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f45abfd119024a9c94814eb89bb88dc6 : SUCCESS in 1h 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005c685972344f25b3e6f0ee1ecc9318 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"09e5a9a00859d78633c69a7ec27c0be6506b9fa8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 13:02:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d8ad86c6bdfeccf5fd7b44af75f4bf68c52410be","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 07:13:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0b79a598fb21d62c5a5c8646de38cd1d3a2ae8e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 07:15:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ffb67a07fb0d295d02d796cdd10edeb25440497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 09:23:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c971996cf154d349fb83e3aad9c87c2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53ea7bce0246473ea10d85d5188e36c3 : SUCCESS in 12m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b97f96e779a345049c112bfb069ceacf : SUCCESS in 1h 03m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d8fe3cab483d4bd7b28125ad48dfcb16 : SUCCESS in 1h 08m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/42d41ded99944c1c9bcb73b546220f64 : SUCCESS in 1h 23m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eba2a89cf31544c98082fa71c5da2b98 : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dffa1f2e8534b4495587e30742b2d24 : SUCCESS in 1h 22m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9f5bf69c5314c51b5a41ca9129748ce : SUCCESS in 1h 32m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dced9fbf3ab14e7ea6c5d55fa3b3e289 : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a366a78fee914b28a7255c31f486bfac : SUCCESS in 2h 06m 17s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ae4c940dd074cdeac16e1321259889f : FAILURE in 46m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e338f511aad74f7aacf163f7999c1410 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"34d1b0e1064230120186d722fb0f7498657d3bdd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 11:12:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77705ff2832e189ce638f4c932776d84e072e326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 14:14:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca14c10e2c6d4fdcafd331faa08ff258\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1f7a8d6d60c41d9b88ebd8f65671baf : SUCCESS in 14m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b61467a990cd4082837bee875509df9e : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d77237293dcd4da6b6bd807c8c6e4f4b : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/31d49e13d3914b03af43c9e58de6a969 : FAILURE in 58m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cb43e7d07ffd4c1382498870ef525d67 : SUCCESS in 1h 34m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a04ba1a1ca14065adbf906b131c2bc6 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/acf71aff5fc94d0985421ec5d7f372d3 : SUCCESS in 1h 28m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06cb2fbd2ed948c0bb6748a16b5bb614 : SUCCESS in 1h 24m 42s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7460bdf20144199a5a9848da0a7fa7d : SUCCESS in 1h 23m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d5bcb2cd91e6499ba529a5d81261bfed : SUCCESS in 1h 54m 36s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/60787cb6eb5b462885dbdf3da0bd99a8 : SUCCESS in 2h 33m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1243c9c7585b410fb36d875987db3a37 : SUCCESS in 1h 48m 47s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee09ff9a041e402b946312af462a265b : SUCCESS in 2h 35m 01s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75ff31ce719e4e359c66c00b7ac5a470 : SUCCESS in 1h 56m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbcdcc3dae5e4af3af6c068b82ba67d5 : SUCCESS in 1h 56m 20s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb7fd6e1ebdf45f6aa8b026b1f1fb45c : SUCCESS in 2h 08m 07s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d564ef64ac4969a015ea360a0082b5 : SUCCESS in 2h 21m 11s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d3f334e0fe0a4d2ea7cc647cc46413fa : FAILURE in 23m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d008b16b28d34a1b9d2a102c7f7a88b8 : SUCCESS in 1h 00m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94d6b50971aa49319736fc21adec297f : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"373fcbb9705e58124f5f517be13468d6ea680a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 14:16:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49115c9b58797952d019baae3a72f67aa7734835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:47:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600284efd17749a5a55c5bb07f66dff3\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5415ce77b0fc4d66bdf6349fd47c5278 : SUCCESS in 18m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/927f4711aa7a432aae79e9a778f7a94f : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b4435b8bf5a41fbad17682d6af9693c : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9b1b3883bceb476c93f764c043ff6452 : SUCCESS in 1h 24m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bdbda39b627494d85aea0008204cc7e : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b908fc6abf2d4a249363d2b464874fa5 : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/310ede2745b44e9ba38297ad0256ff68 : SUCCESS in 1h 15m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b0ab098e4a9486a85e2d9a6770479c8 : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c15ea68e9df472294d345ab3920732d : SUCCESS in 2h 10m 47s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/347255078475407d9db0ec0e4ab0b96f : SUCCESS in 2h 21m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5763cc4efe914b7fa9345aafee5988cb : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3a4fb9fc38aa9c5be7a6a90f61d15804f94765da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:47:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"10720c5c914683d34b3b1127cbdd07e72fb8465c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:48:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40da6d61bac94be38ee64a35d0bcd120\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa57610183048d28295138f3dbf5c7a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c9df6e66547e58e84ffdd7e6c46260a6acd9c87","revisions":{"6df91ecba397a14b319cc2fe1f438426331840bd":{"kind":"REWORK","_number":1,"created":"2024-04-23 16:31:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/916812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/12/916812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/1"}}},"commit":{"parents":[{"commit":"bf25a670ff4a229d7412e85051577f7829168495","subject":"Add worker_self_reference_url to glance configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bf25a670ff4a229d7412e85051577f7829168495"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-23 16:31:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-23 16:31:39.000000000","tz":120},"subject":"Add service policies defenition","message":"Add service policies defenition\n\nIn order to allow definition of policies per service, we need to add variables\nso service roles, that will be passed to openstack.osa.mq_setup.\n\nCurrently this can be handled by leveraging group_vars and overriding `oslomsg_rpc_policies` as a whole, but it\u0027s not obvious and\ncan be non-trivial for some groups which are co-locating multiple services\nor in case of metal deployments.\n\nChange-Id: Ie427f4c69a7bb3c8460400b0ee88e5d482080786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6df91ecba397a14b319cc2fe1f438426331840bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6df91ecba397a14b319cc2fe1f438426331840bd"}]},"branch":"refs/heads/master"},"2c9df6e66547e58e84ffdd7e6c46260a6acd9c87":{"kind":"REWORK","_number":2,"created":"2024-04-23 16:32:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/916812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/12/916812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/12/916812/2"}}},"commit":{"parents":[{"commit":"bf25a670ff4a229d7412e85051577f7829168495","subject":"Add worker_self_reference_url to glance configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bf25a670ff4a229d7412e85051577f7829168495"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-23 16:31:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-23 16:32:45.000000000","tz":120},"subject":"Add service policies defenition","message":"Add service policies defenition\n\nIn order to allow definition of policies per service, we need to add variables\nso service roles, that will be passed to openstack.osa.mq_setup.\n\nCurrently this can be handled by leveraging group_vars and overriding `oslomsg_rpc_policies` as a whole, but it\u0027s not obvious and\ncan be non-trivial for some groups which are co-locating multiple services\nor in case of metal deployments.\n\nChange-Id: Ie427f4c69a7bb3c8460400b0ee88e5d482080786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2c9df6e66547e58e84ffdd7e6c46260a6acd9c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/2c9df6e66547e58e84ffdd7e6c46260a6acd9c87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
