)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~918077","triplet_id":"openstack%2Fopenstack-ansible-os_glance~stable%2F2023.1~Ica91974c587331ad85bdf51200403831a919ed38","project":"openstack/openstack-ansible-os_glance","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-08 16:31:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica91974c587331ad85bdf51200403831a919ed38","subject":"Fix rootwrap.conf distribution for EL","status":"MERGED","created":"2024-05-04 07:09:02.000000000","updated":"2024-05-08 16:32:42.000000000","submitted":"2024-05-08 16:31:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918077","cherry_pick_of_change":917787,"cherry_pick_of_patch_set":2,"meta_rev_id":"ae4174c076b720887e7053836e76095d6320d7f9","_number":918077,"virtual_id_number":918077,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-08 16:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-07 09:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-05-07 09:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-04 09:41:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 18:46:49.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-05-07 09:43:37.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"3e63db015b9c9840caa9fe833ca8651d04b4c464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-04 07:09:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"43e8447775894c74004525eaa8c1750aa81a528f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 09:41:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27336414fe3443909b4f23c99a826781\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3307db2031e54ed880bb0af2fb180667 : SUCCESS in 12m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d4dae407bb30473895eaceb41d452ee5 : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b4cb0e2d1ce4674b7b74b4c24266744 : SUCCESS in 1h 37m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5e63a82dd61443a90cc2f6b3b9354ee : SUCCESS in 1h 28m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9789c61b8d834ff49e79c73b4940c2e1 : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a0c5e72011c418a87b031292f988568 : SUCCESS in 1h 43m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c0805e9efa94314b888fb679b620b2e : SUCCESS in 1h 45m 10s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1e491be29584d9584fe28809a464e91 : SUCCESS in 1h 26m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b35b734a7eb44e3ba3b92afd42c6e91 : SUCCESS in 2h 23m 23s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05208e27027846acb4ce4563b033e7e4 : SUCCESS in 2h 06m 15s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58413a598e624752a799bc6098e66f6d : SUCCESS in 2h 10m 01s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67ab5ad63ea04724af2314fe5065ddd6 : SUCCESS in 1h 37m 57s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c270957ad9704dc08facc5b6049f5f03 : SUCCESS in 2h 12m 47s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a11407729f5e442b8550a4f76637cd28 : FAILURE in 2h 21m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bbe118d87aab4c8ea1ec5d26fd00ded3 : SUCCESS in 1h 12m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/235b2b80ede448e0bdfe74a4d061c448 : SUCCESS in 1h 01m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c12bb5d19534fdaa885d39b72b755b8 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"242c7c0fea934cbb58aa72e43a77e4ec8c17405d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-06 18:46:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa52e21ce646fb5d6807431af4ead79d1a4a3f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 21:14:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed92339e4bbe454c92c7fc28e32fa27b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98366247a2334476b8793146e6bcf8d7 : SUCCESS in 17m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/33fa9e93bdfd40c8862507b1edbe40bb : SUCCESS in 1h 28m 27s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/148d55ade48f45978b797acecd831b27 : SUCCESS in 1h 40m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f10ca8ef512e40ec8c8e51b4f2a68347 : SUCCESS in 1h 40m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c1b1aad019f41fd8d8e1f2b01cc3e01 : SUCCESS in 1h 41m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c09182d303e4580ba0ad459cf3e445f : SUCCESS in 1h 25m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b7095d15fc33487e8549f629248b858d : SUCCESS in 1h 38m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c3e849ee5004f3787e2433294871b27 : SUCCESS in 1h 36m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c6f636f4442946dfaecdb30d3036982a : SUCCESS in 2h 20m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e925bb03281742b696d934b814e8b9e7 : SUCCESS in 1h 39m 32s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4f1738c9ccd4461a74079a4bff4e1c0 : SUCCESS in 2h 05m 27s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc11a4a0ae1f4d88a5d7b6374262d3f0 : SUCCESS in 1h 54m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b501af35b19d44bda52d274ab4c225e0 : SUCCESS in 1h 54m 52s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4500d5100fbd46dca3847a08e8fc9b24 : SUCCESS in 2h 04m 07s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/224a1d36457f423c9b7d92c644c82471 : SUCCESS in 1h 06m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb203d35ac394b55ae5ea6892fef9286 : SUCCESS in 54m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d152d09fd5fc4e48a5e023d34e1aa848 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"555e2d224f0c21b5775dee7b65273f0c3b027711","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 09:43:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3958748814aa79afe8ad5a984fcf9ef77a9bb2fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 09:45:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7ce083d415065e7cadc86fe7a57ffb826c17f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 11:45:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2aa242e351f4f128c5347470d7fcb20\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3792926122bb4965a4effc743f993bfd : SUCCESS in 12m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a5e0e5b0b09c46b8b6a389bd8ed8a5ef : SUCCESS in 1h 24m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b381dd2b4856443684a803b36890ca9f : SUCCESS in 1h 36m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b28534cca7045bd8c4652dc423ab8d0 : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dacefb1d638d488984e0a4590f622063 : SUCCESS in 1h 44m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7726fb63b4d74b089640c9cb856cdf49 : SUCCESS in 1h 22m 11s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/769c2141c0564b6c8fa8928fa1e34ab6 : SUCCESS in 1h 55m 07s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bab3f842dfff401dba882dfe356fec0e : POST_FAILURE in 41m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28321b6b556545ab8842fad5248c8b69 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d0d434888a69780f583386c6cfa9ab96067e6958","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 07:00:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"921e8891bc5a4c063be42159d084713f062e1073","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 11:15:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb4689923ccfd74d261c671a529fbbe7ee018a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 13:42:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c72edb31f42f41b68ebeaae13f2c8dd9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a977535286145a795b1d8c2adbb8523 : SUCCESS in 12m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b196845ec35e415a9f5d4d0ba4cfaa64 : SUCCESS in 1h 26m 20s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/220938a436844b0c9fc8715b329b5cd4 : FAILURE in 1h 11m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/30e2fb722618480687f03d162ace9f1d : SUCCESS in 1h 31m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c8a206168d94a7296e18756019e3bca : SUCCESS in 1h 33m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d13bde19c0e044e9a6c9fc912bd65027 : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/27df76b01670456ab9bbc6049a93b8be : SUCCESS in 1h 30m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ae84d8fbdc544609f7f40dd94b696e0 : SUCCESS in 1h 43m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1fdc7cd281544c26a6305a9236962808 : SUCCESS in 2h 06m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/674b313f8d7542408c3820fc982960cd : SUCCESS in 2h 13m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7aee1e3c726643d9b82fe49a83ca329f : SUCCESS in 1h 53m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32cb36661de44c9586dc90b3cf328157 : SUCCESS in 1h 51m 04s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6acbb5a417664146806f122fc3457cc6 : SUCCESS in 2h 09m 14s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0b1186a8a8e4764bba4f0f11cc9443e : SUCCESS in 1h 56m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ddc2b83711104512897180078687753e : SUCCESS in 1h 07m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7e19824ce2c4b94b95a5e55442075e3 : SUCCESS in 53m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c37a173edea46cb824ace0a0c5d1096 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2015c07a62e9b5fb8604329846a5ae1b4ca45ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 13:44:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69d593b2fd9a7df6e2927261c706758a1b118e2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:31:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a846d597ab14fd4b5f5a485e5e12037\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/531a584548884d34a0de6c20729b1b82 : SUCCESS in 13m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a36a091bd32c4fea9b3ba4f04324a8da : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4f59d050c6e4b7aab66008361c7e67b : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b36da7c4f966490e99e7caa90910a108 : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d8248b9278524f4b911a87ad25b3a313 : SUCCESS in 1h 44m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4a8c3b0f4934a6bb4f411ddcbaaca74 : SUCCESS in 1h 33m 32s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/128114bf084e4da5bd819960cd556605 : SUCCESS in 1h 59m 41s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6228b4c7690d4e6a81b72fedd0b1d63e : SUCCESS in 2h 30m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb76f6a88ba41869f02c383d0c290fd : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"500c40022a48a594642f074c137193fabce890c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:31:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae4174c076b720887e7053836e76095d6320d7f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:32:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e434a2c1a2f43f0b5b67f5eab7b98f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2670e41f43f423b8c845c853f948e44 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5edaa931487d65853ccbf1dd4ba91f236ed55dd6","revisions":{"5edaa931487d65853ccbf1dd4ba91f236ed55dd6":{"kind":"REWORK","_number":1,"created":"2024-05-04 07:09:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/77/918077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/77/918077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/918077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/918077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/918077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/77/918077/1"}}},"commit":{"parents":[{"commit":"a5677359abd57731d1004286e520c10b7c86effc","subject":"Add glance_bin to rootwrap defenition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a5677359abd57731d1004286e520c10b7c86effc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-30 19:36:58.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-04 07:09:02.000000000","tz":0},"subject":"Fix rootwrap.conf distribution for EL","message":"Fix rootwrap.conf distribution for EL\n\nDue to overriding glance_core_files value for specific distros, rootwrap\nconfig was not actualy provisioned for EL.\n\nThis is a follow-up to [1]\n\n[1] https://review.opendev.org/q/I4ee3fc33fdbeb50fc7b102bf62d6134f83c5925f\n\nChange-Id: Ica91974c587331ad85bdf51200403831a919ed38\n(cherry picked from commit 5210cec34b95270d1c3fb2dbc4d7b02654356ebb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5edaa931487d65853ccbf1dd4ba91f236ed55dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5edaa931487d65853ccbf1dd4ba91f236ed55dd6"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
