)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~938448","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I0c197dbe8e0f71f273e3f2dcd9935c78d3044e26","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa/retire_qdrouterd","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-01-09 20:32:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c197dbe8e0f71f273e3f2dcd9935c78d3044e26","subject":"Remove support for amqp1","status":"MERGED","created":"2025-01-06 10:39:10.000000000","updated":"2025-01-09 20:33:23.000000000","submitted":"2025-01-09 20:32:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938448-osa/retire_qdrouterd","meta_rev_id":"eccd42da8fd27263fede56070b7d90c4bcce1329","_number":938448,"virtual_id_number":938448,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-09 20:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-01-08 14:18:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2025-01-08 10:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2025-01-08 14:18:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-06 12:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 10:21:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-01-08 10:21:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-01-08 14:18:37.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"8ef8f38e1beb4e9db42941a6bf6a7c2d32d8ecda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-01-06 10:39:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3ecf95dfe4f40d0ee26659fec69cdb0517f49c5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-06 10:46:11.000000000","message":"Topic set to osa/retire_qdrouterd","accounts_in_message":[],"_revision_number":1},{"id":"ea3bcac88fe1924a06b956a0189407a638e0fa79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 12:59:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31020aef72a943a89051878eefa2b411\n\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/827629de00fd4ac7b823e1db56204cf1 : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/996d70144ba54f8f8b1246dab957ed3d : SUCCESS in 1h 20m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8735a5e442b44510938262674501b3b7 : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40528e00fb4f4a05954c0b959c28517a : FAILURE in 1h 01m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/890e50ffeaca408589387c88578a6a39 : SUCCESS in 1h 23m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51f214c493ad474b8b83ca2a1548277c : SUCCESS in 44m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1de07a8c9caa4317b03c807b00326f6e : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30654de92612431f98cbd38ceb909f14 : SUCCESS in 1h 17m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3ce94169f044a5c8cc6fd55ee68fc84 : SUCCESS in 2h 01m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efd5b0a0f9dc4d6d907b0a5f47a44200 : SUCCESS in 1h 45m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f03020e9cebf4e6f8cd306a9b0e91f85 : SUCCESS in 1h 47m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43c72c2af948487aabfadaca94fe1f5a : SUCCESS in 1h 57m 15s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01c761c6baa24d2e8594367c743fcb8e : SUCCESS in 1h 59m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9269c67b3c5f4c6a9496413bd60c785c : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78cffb9e739c4e68bba83d57a595b2ef : SUCCESS in 2h 04m 39s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f235df116bfe411099cef421c15975a8 : FAILURE in 42m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/075b17b2ac754c98a663a6ff9fc86101 : SUCCESS in 1h 04m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f75b92fcef284d78a97d462c69736e6d : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd8cc31f62a4c10912fdd1c02da9594 : SUCCESS in 4m 21s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ff68cb195c54d288c6e8a8a0f0ae045 : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ef6479dbee484279c9400d3fcc74eed3022cbaa5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-08 10:21:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aea6ca31efd824eb5373223799013c8fa04fc6e2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-08 10:21:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a681768b6067cbc031c9d20998ef1be8c1cb46ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 13:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75c5fde6889640ed8a4ca8fce438e799\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbbb0d7f857c4488b465576a56341d54 : SUCCESS in 7m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f8c45d4359b643bfa5fe11460e0b63b7 : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1d41a30815b4676b49a90cab01a6de5 : SUCCESS in 50m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a094dd93a7894281be009ba40f85bb65 : SUCCESS in 54m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f93050218704aa2bd48efdb9e9f0d18 : SUCCESS in 50m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/713bf53708a9471d9d9e5fbc0c49189a : SUCCESS in 53m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e7d36a91f1634027a14dc1aaa18b0413 : SUCCESS in 1h 28m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f24bbe1afd494f3abef14c64ea0d9516 : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/debdea8990c941d38518b8d065c74250 : SUCCESS in 1h 22m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ada02e51f176411dab32f498356fe1ff : SUCCESS in 2h 02m 37s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/40f62e92a3dd4661bbd38c78f6c27a10 : SUCCESS in 2h 01m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d408f2de2f8842faaadf6d7e55255e7a : SUCCESS in 2h 01m 29s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/979477bc4457493bba7aee245dbe5d19 : SUCCESS in 1h 51m 59s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa55db3d52a94e92b10bbfd6c267dde9 : SUCCESS in 1h 51m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d5a0e3275cb4921a68e543ac66fe385 : SUCCESS in 2h 01m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4360521ab19049228dbda8de75f42d70 : SUCCESS in 2h 05m 38s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f384e7d19ba641c3829cfa474a3bad69 : SUCCESS in 2h 04m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/978f0066e13d4d75adb7581a267d58c4 : SUCCESS in 2h 09m 15s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/09911dc3b0c74dd2828ca43230208239 : FAILURE in 49m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/059c656dcfcc4ad6b0378c1fcbefd55a : SUCCESS in 57m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58e7568cf0a5433c8559e7d559156286 : SUCCESS in 1h 06m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2437ba94df8741588fec727ff0da92fe : SUCCESS in 3m 51s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51d907b8ae7e4f619759c3e6d1767668 : SUCCESS in 1h 19m 19s","accounts_in_message":[],"_revision_number":1},{"id":"514fb84967ed893d3875316dbb68b439976ea591","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-01-08 14:18:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95684b04edca136f5e74efa3ae53f16e46f5f3ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 14:20:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc56787723153810512d52000490e7feb9397d05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 17:35:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae71d2a4cf6a41fb91993f8c359cdcbc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b045b75da41142f4a67b77dad15f95e3 : SUCCESS in 5m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4bf4932622244c6d89ff484d7eedd36c : SUCCESS in 44m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/560677f061184fd9a921148872d61a7c : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b62471008014de9a11fbd350a92db21 : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd890a5332ee48f8893fb108bc286a84 : SUCCESS in 1h 12m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc6f59cc1659455985a50af360b6e9fd : SUCCESS in 1h 15m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/040e628a87c84dff8635805b7ebddc05 : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d741fa5c59546d6b3a403c263a3df1b : FAILURE in 1h 05m 57s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f97835cf49b45749f7d1b1186ff5a25 : SUCCESS in 1h 47m 14s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82b676a782234f09b0fb1904e65e3489 : FAILURE in 3h 01m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/040396773ee44ebf94a0c80df610089a : SUCCESS in 2h 04m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1d21bebe1747cdbcf386e23d36ef43 : SUCCESS in 4m 27s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ebbd67ad52a4d69932a4477ab3fccac : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":1},{"id":"83cd06d0989bc250ae2ea22ed349edf424d5b3fe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-09 15:55:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9176f795ce46d6d9ca1373ebe17055f37cda8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 18:28:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b3cdd40735549c9bc1a55d88f02a3e0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d00279c083942c186e2b05e6e370a58 : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f87a189ccfa43a5932a94972cb6b7e8 : SUCCESS in 1h 21m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4e3c64bf4ff41ef92790d2bb9a475b7 : SUCCESS in 1h 12m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/879bc37385e84adc8ae5ed4a1d3391b3 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/123c3488943a4e96bceaf984b01c28e8 : SUCCESS in 1h 12m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69fe8511f6ad40c382cc70f7407d7d59 : SUCCESS in 1h 30m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/697a7015acf74a48bc6f57e3a5e8a9f2 : SUCCESS in 48m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2048c23df8d54846bc908eb3bf6f5b4c : SUCCESS in 49m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/303f4205254443249f80c0debbc2ee10 : SUCCESS in 1h 17m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc40c0b2f943417abe6b4ab14e04a882 : SUCCESS in 2h 04m 45s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6140b998d2f3409982712966c7084b16 : SUCCESS in 1h 08m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/892b67ff2ef54467a9adb819152ed3d0 : SUCCESS in 1h 42m 14s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a2c4b373cbf4f1c89efb34c8705b511 : SUCCESS in 1h 43m 45s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/09a3473c55cf4f1ab94528decf1801d1 : SUCCESS in 2h 04m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddcda88488894e7489cbe2412e15c193 : SUCCESS in 1h 57m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea11a2e2de41486f8079c77673ffb228 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9800c861c18a461a9e291dc09099eb56 : SUCCESS in 2h 01m 31s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2275aca93ea4e0590e1bda5f97a774a : SUCCESS in 2h 29m 49s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59c2743629794d4eb58c95d708511a7e : FAILURE in 45m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8dc49ad3999641fa8c1b0ee8b38a784d : SUCCESS in 1h 01m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da51d796302a4d7e9b4f13e81b54a587 : SUCCESS in 59m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57db87980de841b195fb302801c2a62e : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/615290aaac4c4d46bdcd8e8c0258e9c0 : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"21bb3ebe80791b9bdd89aa196f7833518aad581e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 18:30:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0743c610162d4b7255923660d644de7834241c79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 20:32:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738b2ff378ea4a3b8c871405a4a09932\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef20de424940437e8cc4da8be94bb635 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/770324e5d2554d508108382ea2f36de7 : SUCCESS in 1h 04m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/526a3f1f01dc42cb93cb4594e25d060d : SUCCESS in 1h 07m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e47dce6452bc49e081da5798d03a780a : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4b5d980d71740fab40101b0485fcbf2 : SUCCESS in 45m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9b052beca8c4744ba009374bce53af9 : SUCCESS in 1h 21m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/611f54e8eb7a4935917a47f2af10fe69 : SUCCESS in 44m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d12dd8af60cb4a989d5be10906ccbdc7 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99a234d5f79b4647ae74f32b072f639a : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec631ab6093a471b998131ca7c42a6ca : SUCCESS in 1h 54m 50s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30bd484cc86144aab82935384080de2b : SUCCESS in 1h 59m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8edfb63fb341949a6e1e9015029252 : SUCCESS in 3m 58s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/784a8ebd28e54007a80f85b862cdb34f : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4ccbec8c593fedba77debee0f0f8394c66c3b7d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 20:32:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eccd42da8fd27263fede56070b7d90c4bcce1329","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 20:33:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f73e55102aa4d5f9515712c36ab369c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce9d994673b491f8f3aff45eba0b125 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f72790b4cfcfbec940aca760ddf017b421ec981","revisions":{"8f72790b4cfcfbec940aca760ddf017b421ec981":{"kind":"REWORK","_number":1,"created":"2025-01-06 10:39:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/48/938448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/48/938448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/48/938448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/48/938448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/48/938448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/48/938448/1"}}},"commit":{"parents":[{"commit":"96cd191579376945b98adc548afc73e8f8330972","subject":"Ensure that first/last host detection is deterministic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/96cd191579376945b98adc548afc73e8f8330972"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-01-06 10:38:56.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-01-06 10:39:07.000000000","tz":0},"subject":"Remove support for amqp1","message":"Remove support for amqp1\n\nSupport is removed in oslo.messaging so we remove support in\nopenstack-ansible roles.\n\nChange-Id: I0c197dbe8e0f71f273e3f2dcd9935c78d3044e26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8f72790b4cfcfbec940aca760ddf017b421ec981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/8f72790b4cfcfbec940aca760ddf017b421ec981"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
