)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~949428","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Idc4a00ca25510353d757bd13e8d815a38a5e7d90","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"osa_hashi_vault","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-03-29 12:33:37.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-03-26 21:08:41.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Idc4a00ca25510353d757bd13e8d815a38a5e7d90","subject":"Add hashi_vault pki backend support","status":"MERGED","created":"2025-05-11 20:34:58.000000000","updated":"2026-03-29 12:34:27.000000000","submitted":"2026-03-29 12:33:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949428-osa_hashi_vault","meta_rev_id":"790a851c3bfc0c1baec12a10d590789d01c5dd8c","_number":949428,"virtual_id_number":949428,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-29 12:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-03-27 12:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-26 21:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-03-27 12:12:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-11 22:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-10 17:49:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-26 10:09:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-03-27 12:12:27.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"2fe62e4f5737451dbb410246b87fade43ffb96ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-11 20:34:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5204423057878c4b48a527b7f966935c8855c783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-11 22:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a815df15afe142e5b1d479e80b38e9ff\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb67dcf8ae114e0ba7efd0b55b97decc : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/462983eded21450db211b818a9cf0d73 : SUCCESS in 1h 08m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/84162838d5b14c2c9a58d8f3f382d9df : SUCCESS in 1h 12m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75f9dc882fa14408aaf3274f1feb6cf5 : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fe5a1ee19864c9eb39974ec0313a0c1 : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af12c24823a24a27b98eca4d91904743 : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d2234b427f824216864f6c118c05da8a : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52af31bdd4c44d74924a59107250a4ad : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58edef7135284e179d652b1d5d0c7c24 : SUCCESS in 1h 29m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67f938b2ba5f433aa328bc8ab030a509 : SUCCESS in 1h 57m 00s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d98e6ebb5ca485ca287045168424c23 : SUCCESS in 1h 47m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c487edc0d044c658f3f165d7cf97bc5 : SUCCESS in 1h 47m 24s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63946552d9424d1586becd01776d223c : SUCCESS in 2h 05m 38s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4668ba27ebdb45bc851f956a5995bbd6 : SUCCESS in 1h 49m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/161ca46480aa4fbcaaeb67ce825326fe : SUCCESS in 1h 52m 52s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ce374f417744a2db4183d52ee836267 : SUCCESS in 1h 54m 22s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fea33501d8e6418ca7755786608e7a56 : SUCCESS in 2h 03m 41s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9db49ed03b4742b887939bbc1c88c1d3 : SUCCESS in 2h 01m 53s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d0c95555d1644729f528a75a2a2a4f7 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87d25f8d65ab43f1853ad72fad417830 : FAILURE in 43m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32c9d2f4b56142f4bc72acbd63cf8dca : SUCCESS in 54m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8870f4b3afb04425a050b2d7a0a92a32 : SUCCESS in 56m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca87c8ecb6cd4b5fb2cb1ca86881746a : SUCCESS in 6m 13s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af77a3e1460d488eb9df21d0d64bf3eb : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"92c04270a422ac164826504321be2c5fb12d880d","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-23 10:31:50.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"5b49e7c7e47f3541c13e0776ae9284ab54aff828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-07 19:41:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a03d4fb702f864b13933767689b6cfb5b93e5de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 21:32:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9fe9721a20a4165af6770a1c2e9c3a4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed2aec7b398447d6a753429b600a9dbe : SUCCESS in 7m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d3cb18c30bd94127863c2082aca62b71 : SUCCESS in 1h 04m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/795555eeecf24eeeb84a6e7e9764bf9e : SUCCESS in 37m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73f0b448d5e3455391956a9210fefc77 : SUCCESS in 1h 15m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/83e2eb9f2c0c42f9ba87188fd0c8f1ca : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b18c88ac5dc5479e97a7996d0aa3b9a4 : SUCCESS in 1h 14m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04622c839fc44b2ca67eb80ad5b16f1d : SUCCESS in 1h 11m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6b8d60bef7549cebafaf6ac36ff4a7b : SUCCESS in 1h 45m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8be76622de4d4d2294d459b0314f521b : SUCCESS in 56m 19s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae0d807daa0a45bfb653a05c2e32040e : SUCCESS in 1h 46m 44s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e61798818a134ab79c2b40e556e862a4 : SUCCESS in 1h 11m 22s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c06b28f40f9d40d584307c4c0b5a0d6c : FAILURE in 8m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/befe9487997a4d0bb7d093506b2edf5d : FAILURE in 59m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ae316c5a7346b1bfe8cd2761034b81 : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaa3a0eaff7445d290736aa583898ca3 : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"045463253ff4db918bbd2feee0ebcc505d171a83","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-10 17:49:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"336e8c87280ed58c057c18ea5c73da32caf14a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-13 11:48:45.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"42f63dc2b301c75ec8b923bd5b95bf7b9e3687ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 13:57:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c1fc97d3ecf4b74a54f7e9614c7b04d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0edc8fb4b7c44bdc8fe5dfd13188d336 : SUCCESS in 11m 24s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/eaf12984d90c4946967b26e749288dac : FAILURE in 1h 02m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3c8cd23417b344ebbb883b3fc1320f9e : FAILURE in 40m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ace19cdbc734418fa4dc41a9e2106303 : FAILURE in 1h 17m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a5c7cb9920f0446d897e3585d50d6300 : FAILURE in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6491e13d030442a5841b3eb2208f9c50 : FAILURE in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b683385ea9748b2bb7ad0f03c3de053 : FAILURE in 44m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4f40794e5f94f998612a2a6c1c50ab8 : FAILURE in 1h 17m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d265a0ac1b114a64b93a6da58d51ecf5 : FAILURE in 1h 27m 47s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ced98a72a18c47c0b277ab4e145ac25c : FAILURE in 1h 02m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/103c40e9c295486a939980359697b1bc : FAILURE in 59m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/726a08b712bc4d06ac1597451e1d81d8 : FAILURE in 1h 32m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/939d366c9d7345ab8e8507617355e9a8 : FAILURE in 59m 45s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88e74ea815de4160a88dbce5e8651a96 : FAILURE in 2h 04m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf400b3cddd5405f8855804a9a0221ea : FAILURE in 1h 59m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a16daa2a7b254eddb7473aed879f1160 : FAILURE in 16m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c203ace402d149b7931a3e2cb3618ab0 : FAILURE in 29m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0dd41379644461d8deac2c539431f9e : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/133ad153f7c247ee9e589f049590defa : FAILURE in 1h 21m 09s","accounts_in_message":[],"_revision_number":3},{"id":"48b23f1fc8504061621168e2a0ecf467805a51c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-17 15:11:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_32666\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"_revision_number":4},{"id":"3f553673f43fcecbc985764928ad8c798c96d106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 18:14:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0797649676534cd181141a0b760462fd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac1387f0ae2145028b853a4e97bab00b : SUCCESS in 14m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0bbbba80bde54487932ddf6c1a893b51 : SUCCESS in 1h 20m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/37ee5462be7043a8abd8b61400b66ef6 : SUCCESS in 1h 46m 40s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/aa2e314cc6a84bba8a360ea271733b89 : SUCCESS in 1h 38m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/515196a63f3b4bafade60880c6d6f309 : SUCCESS in 58m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/78d1369ac53b4878a9e4a5d298c2a190 : SUCCESS in 58m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/147a4f8de6e145f5a8da78437812023d : SUCCESS in 1h 33m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e7172419a6d4d50a1b2644fc3aceed2 : SUCCESS in 1h 42m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/386ee60264b540278859cd66bc7d7fb6 : SUCCESS in 1h 21m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c69c51d364c94eed946427d9ef3ae2f0 : SUCCESS in 2h 37m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e1bc571341044c6a39344b752832ab1 : SUCCESS in 2h 31m 38s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e305da105ba84564b62127dad28e0b62 : SUCCESS in 2h 45m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b50109710c045eca65d060b44c87531 : FAILURE in 50m 59s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/784a3da9ea284b218ec6747822d86c23 : SUCCESS in 2h 50m 35s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fb70aa702974349a7bb379d4029b2ad : FAILURE in 1h 21m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/49c5731efbcf4042b018e6afff6dac8e : FAILURE in 13m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ad6ff141684410ea8cbb9aaa025ee8b : FAILURE in 33m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01dc6756e47b48399fbf623a687ca161 : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/193ea1bb080e4e278e8e08bccf8f2b53 : SUCCESS in 1h 42m 57s","accounts_in_message":[],"_revision_number":4},{"id":"b0d7cc02a0f07ef64449a5fbbcbc51c349684fba","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-17 20:03:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7313fe033f90700ff3c23c6aef575ce5d9c0a14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 22:42:43.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d8d1f0f4fa44a94b41312a2a75c8592\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f77926dca6a42a08b226f8963a01298 : SUCCESS in 11m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/21f36a2173584dd18f3405b02eda34ab : FAILURE in 53m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bf2e3e9744c44839a9a544320aed8e16 : FAILURE in 1h 16m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/702188b438e94955b7ae457bcc146e75 : FAILURE in 1h 16m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a522634309a94d2089f0845bf91e9ddb : FAILURE in 49m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/521141c78f254c5fbeac0198577074fa : FAILURE in 55m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75883cf43b4c479c8e5b6bc4a387e4ab : FAILURE in 53m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6387c45566f7406183b1a26c62b84368 : FAILURE in 1h 32m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eab6d9850225430f8f98f7989a5d5b40 : FAILURE in 1h 13m 34s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73f1330ecf2843499026614847f937e6 : FAILURE in 1h 38m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db79113c0fe54331b01911f547fcec02 : FAILURE in 1h 02m 35s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93a2747a809d441fbcc8fc3babb2ae88 : FAILURE in 1h 09m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5bfffa4b3434e24806a320b499e1eb1 : FAILURE in 2h 16m 29s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9658a9ea8727480d97ba64d27547fe77 : FAILURE in 2h 08m 49s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/063b5f3155924e3c96bffb50a50b7909 : FAILURE in 2h 31m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2786df0b25034372a6c941f9e1e3aa72 : FAILURE in 11m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d55ed5fb10a4f7e8c493011bbae1ffd : FAILURE in 1h 02m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6e3f05f24441fc90eac047a1c99458 : SUCCESS in 6m 33s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c20858b0830c47a5a572d64bf3f9652f : FAILURE in 52m 58s","accounts_in_message":[],"_revision_number":4},{"id":"8c664a854df51632db1be4c4a1c8fce15e85cf71","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-18 10:10:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aec40099d96f04a236c1248322e61a59ce76e16a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 12:57:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30741ed100224fec9d3c722f5e989040\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/854a5b1a6d9e45d1a4e369ce31221396 : SUCCESS in 10m 24s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/65f6448b9e9f452989abf828404e0758 : FAILURE in 58m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/87ef52c6ad8145c98497aa44a469f9bf : FAILURE in 1h 18m 24s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a3ceec0d28e245bfabb8ef353c0b7d04 : FAILURE in 1h 21m 23s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/57f9de204900499395f87fb660e4d98d : FAILURE in 1h 34m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f2d53d7877684849b370edf60bf0cfad : FAILURE in 1h 25m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/953ac2185c334e1b9df716dd1f9ab306 : FAILURE in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/049bda35b40a47ff8bd714d5f2e8cb70 : FAILURE in 1h 26m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8d6e9974aabc416a8fff16b753de3b15 : FAILURE in 1h 15m 57s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/50636c1041b64a889fef795d6036589a : FAILURE in 1h 57m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53bf95bfacb5453297237958c4efef6a : FAILURE in 2h 06m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb40655a9b6b434cbb8aa3127c9294c2 : FAILURE in 2h 05m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/817e6d2d9e89461c8af4111b0e22b2f3 : FAILURE in 1h 45m 40s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6cb42bb1e84460eaa89c7600fcd6cc9 : FAILURE in 2h 39m 27s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f86238c3e404413979f5cebb7ba2e36 : FAILURE in 2h 38m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/47321a4bf1ec4c46b7bd992972fc18b5 : FAILURE in 9m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b37fe5fd92924ce4ada13741b110263f : FAILURE in 50m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba05b3eed1647c69f6c87fc6ca2c9df : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90bf150ec7a44f4c80976b6875cec556 : FAILURE in 1h 13m 31s","accounts_in_message":[],"_revision_number":4},{"id":"a175b81a48033195d6c4e6bf4658361f1087e2f3","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-24 12:48:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"638c41e433e9febfb6b45bbb3ff62ad307312c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 16:01:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9719a57bc2004cd7a89c8f79610a91e1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f517cc53d40a4b558b454842034b0476 : SUCCESS in 8m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d356b5150d0a4fb1b756a79a0e5bc1d9 : POST_FAILURE in 1h 21m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/100de31f8ac8425da621a74515affb6f : SUCCESS in 1h 31m 49s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/5a56866a2e9c493db376a4ab56477c7c : SUCCESS in 1h 34m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d4934f9b14e419389b1276f4a4d0a66 : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ec1ca8ad1ce745cb81dedf883acf79e9 : POST_FAILURE in 1h 45m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc8f0fded86d4453960443fcbb970fbb : SUCCESS in 1h 39m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98ac63d5705743bc816f099b16d3fbfb : POST_FAILURE in 1h 42m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9afb14f57f16467fa92a0627b3f5fc00 : SUCCESS in 2h 23m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e255f497adef4eba9939a5023659cc17 : POST_FAILURE in 1h 13m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9952f8118c374c0fbe7d6ca8a2f73fda : POST_FAILURE in 2h 20m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/812308edba1145cb9ba5b05bcd261534 : POST_FAILURE in 2h 17m 51s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d31733ca9bdb48e580b6eede964d7a15 : SUCCESS in 2h 12m 34s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9797ad9e7943420282a2fac19b596ae8 : TIMED_OUT in 3h 05m 02s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c400adc829e3401d9217b5a52db246df : TIMED_OUT in 3h 05m 53s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/29c9ffdb47034fafa320fa5faa62f9c4 : FAILURE in 7m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c9100295f224eae82f2deecad54e4a1 : POST_FAILURE in 47m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6232a336f6a347c8ba08e287618a80d7 : POST_FAILURE in 6m 41s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb99d6989ac44bfd950d57047b4c1698 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":4},{"id":"63b5094f972cf3ee30f94aad981a72b3bfa96bb5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-26 10:09:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f14e29bb240df97108442f2db99fdbbbbba0eb7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 13:33:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fad1f903b7f41d38ec38aa754e90089\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a872bec6682b4aa1a15a96a72aec3ec7 : SUCCESS in 13m 17s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/22cf734acfe4458189d0a355db84b0e5 : SUCCESS in 1h 41m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c94c230252014af3a3f960a725d2507a : SUCCESS in 53m 05s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a1569d99f0eb4329a81de7d6d3d126fb : SUCCESS in 1h 34m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/afbc942ca84a4d3eba8e16bed1f1a20f : SUCCESS in 54m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/91a4cb5a703c47048a151b4b8a8726ad : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd651cac95eb4df8be26b9808b76c9ca : SUCCESS in 56m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73c2833c097c4207a4241aac5ecddd01 : SUCCESS in 1h 43m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/60e2ea5e69594d62b1337b52fb6f8fbf : SUCCESS in 2h 33m 24s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/13c44fdac9584e9eb735953d47e3ec8d : SUCCESS in 1h 13m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c89554666b5a4b0eb3a00ed95b03b839 : SUCCESS in 1h 13m 10s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f24d7383340542f1a48e1623221e6d4f : SUCCESS in 2h 12m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9a23aa6d91c4669ac859569a24f3583 : SUCCESS in 1h 10m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a75387a66bd42239fafd5493005caae : SUCCESS in 3h 07m 20s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f632e523e1149098820df55d5c1f3be : SUCCESS in 1h 33m 55s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2b478f1b6931402daca25137fc5a8cf1 : FAILURE in 12m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d01cdfa184484b1c96309f99ed734e0b : FAILURE in 50m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974649f08293423689fe2e04fe72fca1 : SUCCESS in 6m 05s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ea98c971a394c74b3d945bda79dbf16 : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":4},{"id":"cd21b6fcd31ce4d2bc9499b3bc012b7afabfd952","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-26 21:08:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c5060576b213c6a2375ee89d8fd236e8643e3cd4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-03-27 12:12:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"990406ed1b5eb08e664f591de8ce6f674051c9ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 12:14:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6ab892ff1fac74460b0f589521d29447ef7a95d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 15:30:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55d49e6f03544c859829f20d394c52bd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2879151cd19f4673afd733e58d1b0cdd : SUCCESS in 8m 32s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5d656508f2224a478c9a5d24e8c8ae05 : SUCCESS in 53m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5712d9c47f3a4d49ada4d3930d04cb5b : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d5af498593454eebb50b831a8debae2c : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b00ac3c96f9844088af17bfbc61d1c20 : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d6c5bd49cc74bcb91141b9357533a1d : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed71a0025d074f3d9f27c69740f46185 : SUCCESS in 1h 55m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a454c8050b945cbb314ce3b97ebb88c : SUCCESS in 1h 43m 41s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69302a28496642cbbe36b24a68e52774 : TIMED_OUT in 3h 10m 48s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8c3a42137b6421e888f114c4fccf25a : SUCCESS in 3h 07m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c4eab3645b4179bec914a5be90d1ce : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0598758b64ca4459a64948a0789c6733 : SUCCESS in 1h 43m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d82212a40af9b29991e94eab79c224c8312826ea","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-28 08:31:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3081d7d89026a8a420cbfb71eaf9359e61298e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 11:22:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ca7e03eb824136ad11511f81059086\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61c4098f9078437fadac7ac35f3ff51f : SUCCESS in 14m 12s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/de100f0758ec419f90858b9763fba201 : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/971c29a582f44679acb86210c74fbd32 : SUCCESS in 1h 38m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/4ec1e1c494634c57b44d538532584986 : SUCCESS in 1h 30m 51s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1d09cca99de243eaa47366f0bcfafcb2 : SUCCESS in 1h 00m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/55623f3633774120b1a27cf46f6c7fc7 : SUCCESS in 1h 22m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcead5a43e8d45cf8684e28a61b133e4 : SUCCESS in 1h 36m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff96b754233e4c50ba7fb56d6d565047 : SUCCESS in 1h 41m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ff9e0b5e0f6c41fdb69ca0578bb7b5de : SUCCESS in 1h 12m 59s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/01fe5819de17407cabc43d69db7b3bce : SUCCESS in 2h 24m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ce98716afb94197bd4b017577bf808d : SUCCESS in 2h 16m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e742c8b831b240fc9368c89043ca3e36 : SUCCESS in 2h 22m 56s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a398b6ba5f34107badb3e6f3b8b059e : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f04b06ba4cdc4ee89449ad0321c17434 : SUCCESS in 2h 32m 44s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/defdbf2912c24c51b47339874090154b : SUCCESS in 2h 36m 59s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/782a0f01e7f0448ab93e953b2fbe3951 : FAILURE in 15m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8d72fcc1336447f94a3ec4bda1a5427 : FAILURE in 49m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344ed94ba04240d2b2b256f890658d2d : SUCCESS in 7m 11s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/140289e315794d4bb8ec0ee602f765ec : SUCCESS in 1h 44m 41s","accounts_in_message":[],"_revision_number":4},{"id":"7be58671f66f5d0e60e87e329854100086e8f375","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 11:27:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"75647485152dc83b186827d074b9f1e81d6ba367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-28 14:16:50.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/949c8b0133e94b699d6c9a52301e29e1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99e06b31f30b4c92904998bfd1bffd75 : SUCCESS in 10m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2512ecf885fb4e9182b9534e13abbf0a : SUCCESS in 50m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/04ddd56902d14be2bda30ebc379d3d2f : SUCCESS in 1h 31m 53s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d2fd24f24d0841ce9ac30181985ce036 : SUCCESS in 1h 35m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bb8f64eb9d2c419698c8e8ce34987445 : SUCCESS in 50m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e2fddbc08a6948ab83485ccf73726a5d : SUCCESS in 1h 44m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc981b0163da4d41949255dff34783dc : SUCCESS in 1h 33m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bde7b26ea76d4e6ea1d993ba8b4c50af : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebe1e2c2911b4206bb6352f3a17acbd8 : SUCCESS in 2h 44m 42s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e9dd2ecd0d2487d9bb31cd7ccef6cc8 : FAILURE in 1h 17m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548c5c4e0e5642c4b8fd985af00a5b60 : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c2ebf3e89a74f3fb3646f1c0d8c4213 : SUCCESS in 1h 36m 28s","accounts_in_message":[],"_revision_number":4},{"id":"21c6f9a1c9743e5b55519d7db8a7e55c45a78f6a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-29 07:02:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e85b71a620dd2d4b136622d76cb8f8e405ed4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 09:54:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ffd758224c748369ed99967d8164ecc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4123ea9669ae4198b77ddd9b0a5bdaf4 : SUCCESS in 18m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a70e8072ef8d4d3f9e3e7acbc3affc3b : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5ec56ef517cb4a1e87be4afc0680d983 : SUCCESS in 1h 34m 01s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/687fe58d000642b6888e0acc6760f042 : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/853791ad859544eb8abd7812be4706fd : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c933ddf856994e5eb80a918237a27c4c : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb45f98158be4ec59815d56ad2201453 : SUCCESS in 1h 10m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53d6cd5aec724f50b298ca0838268ba3 : SUCCESS in 1h 45m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3735d2f0aed047db90b84ef8990f9f30 : SUCCESS in 2h 32m 01s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b3a73522d094f1786e556f62231e678 : SUCCESS in 1h 41m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f8eec01c67b4904b7baba63abcdf050 : SUCCESS in 2h 29m 03s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76fa446f08a64be5a3ec88fde12fc2c7 : SUCCESS in 1h 58m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46813352a7c140e2b95d855de00d8d78 : SUCCESS in 2h 12m 35s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2687fe353c241cca412bc086336a73a : SUCCESS in 2h 42m 37s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0beb184971fa4013b8e4769f3c55aae4 : SUCCESS in 2h 16m 11s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5664e3dc083946ad94cdb7712a63d810 : FAILURE in 9m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee75ec0b7c6345c89e8b00e160fdf880 : FAILURE in 59m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847e495596e4472f9669e0ed20ed4f71 : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/832ddd825c2a473da3521d49c07b636f : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":4},{"id":"c0e3fcfb8ec168fd300fe3ddef987627f132f117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 09:55:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d4261cde6499c1c457a41ad189e95528dcf7f933","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 12:33:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e77490fb406c4bf290918281fab040ca\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a909cc863a274fd1b59cb82d900cbee9 : SUCCESS in 7m 53s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/98dd36e916e545e391bd40d2bb4f5d5c : SUCCESS in 50m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6f365c3b6f734179bf0fbc5e19ada304 : SUCCESS in 51m 59s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e3353f57042c4566a8d8e1f052a589da : SUCCESS in 1h 13m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d7febf2b2a34d638531030f35867980 : SUCCESS in 51m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c8cf0f8d95fc48a9abb8addce651edc8 : SUCCESS in 1h 25m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e6df349a9d14d4aa2fe2286681fd72a : SUCCESS in 1h 36m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a817826514de471c9d03c6001279d7c3 : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9ef41c2b71c424994543c56a9563ff0 : SUCCESS in 1h 52m 06s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/497db3f2b86841cc928c1d7c173656e0 : SUCCESS in 2h 33m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da219d77f86143ddbad73deea94c4192 : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4042b91404de442e87683d9ab62fe267 : SUCCESS in 1h 12m 35s","accounts_in_message":[],"_revision_number":4},{"id":"803e1d3615ddbfedf56ebf62958a3676cda226f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 12:33:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"790a851c3bfc0c1baec12a10d590789d01c5dd8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-29 12:34:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb723567cb94ef1985e7ca996bad84b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75dcc54c47224480919392deeb5b7fa8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cba90137e828324b6b82b3c87616e6ee60f17516","revisions":{"1f1dc604b1cf0c5b4c2cd8c502299399e6848b9a":{"kind":"REWORK","_number":1,"created":"2025-05-11 20:34:58.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/28/949428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/28/949428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/1"}}},"commit":{"parents":[{"commit":"c5ad57d456967dd189cb130fa4315f3dd11a1a1f","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c5ad57d456967dd189cb130fa4315f3dd11a1a1f"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-08 16:51:55.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-11 20:34:56.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `glance_pki_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n\nChange-Id: Idc4a00ca25510353d757bd13e8d815a38a5e7d90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/1f1dc604b1cf0c5b4c2cd8c502299399e6848b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/1f1dc604b1cf0c5b4c2cd8c502299399e6848b9a"}]},"branch":"refs/heads/master"},"e6f91010bd4cff88060b610c96513b85bf75a463":{"kind":"REWORK","_number":2,"created":"2025-10-07 19:41:32.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/28/949428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/28/949428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/2"}}},"commit":{"parents":[{"commit":"c5ad57d456967dd189cb130fa4315f3dd11a1a1f","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c5ad57d456967dd189cb130fa4315f3dd11a1a1f"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-08 16:51:55.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-10-07 19:41:19.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `glance_pki_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nChange-Id: Idc4a00ca25510353d757bd13e8d815a38a5e7d90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e6f91010bd4cff88060b610c96513b85bf75a463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e6f91010bd4cff88060b610c96513b85bf75a463"}]},"branch":"refs/heads/master"},"23bf0b6abb2c3f05052b5e10473713aa643e0f5f":{"kind":"REWORK","_number":3,"created":"2026-03-13 11:48:45.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/28/949428/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/28/949428/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/3"}}},"commit":{"parents":[{"commit":"c5ad57d456967dd189cb130fa4315f3dd11a1a1f","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c5ad57d456967dd189cb130fa4315f3dd11a1a1f"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-08 16:51:55.000000000","tz":120},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-13 11:48:45.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `glance_pki_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nChange-Id: Idc4a00ca25510353d757bd13e8d815a38a5e7d90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/23bf0b6abb2c3f05052b5e10473713aa643e0f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/23bf0b6abb2c3f05052b5e10473713aa643e0f5f"}]},"branch":"refs/heads/master"},"cba90137e828324b6b82b3c87616e6ee60f17516":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-17 15:11:39.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/949428/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/28/949428/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/28/949428/4"}}},"commit":{"parents":[{"commit":"fe0a75cded8234aac6bb22097c9365cd16f0a4d7","subject":"Merge \"Remove API v1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/fe0a75cded8234aac6bb22097c9365cd16f0a4d7"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-08 16:51:55.000000000","tz":120},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-17 15:11:39.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `glance_pki_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nChange-Id: Idc4a00ca25510353d757bd13e8d815a38a5e7d90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cba90137e828324b6b82b3c87616e6ee60f17516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cba90137e828324b6b82b3c87616e6ee60f17516"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
