)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~949653","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I0728383786d6f2282888fc1bf6ddace7773f2e40","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-15 07:21:21.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I0728383786d6f2282888fc1bf6ddace7773f2e40","subject":"Use dynamic include instead of static imports for conditional tasks","status":"MERGED","created":"2025-05-13 16:27:18.000000000","updated":"2025-05-15 12:35:28.000000000","submitted":"2025-05-15 12:34:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949653","meta_rev_id":"9fc6d0824977b2047eab1e206334ab9e7388d829","_number":949653,"virtual_id_number":949653,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 12:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2025-05-15 03:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-14 14:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":1,"date":"2025-05-15 03:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 18:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 14:18:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-05-15 03:12:03.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"2bb4c1227921c1b46bae95ebee1bcce017d28904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-13 16:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2f636f2c4f63b881ba1341793d532504b402dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 18:33:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d436e0366734e0a9f108c19f0ffdd54\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/050001b29d034c50a539308bc20413fd : SUCCESS in 4m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8ad59df479f841548b795ae2811d2f2c : SUCCESS in 41m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/03e4d16ad9ea4992a24cd1dab33506d6 : SUCCESS in 1h 26m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/992abbcedc1c4be4bc3cab94ad4f7b31 : SUCCESS in 39m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e42ab09a76f4a59afdd2c89dca39279 : SUCCESS in 1h 09m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d088eb9038154c168d4e07a25c0bdcb6 : SUCCESS in 1h 12m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8380eaf332a843f180bc3320590d6b00 : SUCCESS in 1h 16m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93de4eee99fe40ebaab59eb955e2daa1 : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/800510a1954a4810930290d76b63c6e1 : SUCCESS in 1h 14m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6125e4056a994b1faeaeae21e84b10e9 : SUCCESS in 1h 38m 50s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2dc743bd518d4f70a48b8aeb2175963f : SUCCESS in 1h 45m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b8992526eb34d0e9d7cb8e9d6b20c18 : SUCCESS in 1h 38m 12s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff940138d5f5404a8250f7b63fb4cacf : SUCCESS in 1h 34m 19s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f17766a07f345b5813b20a2d52a1ade : SUCCESS in 1h 38m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/803c38a26dfe470cb9d9545a40358dd6 : SUCCESS in 1h 43m 08s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e64e01fcb758400d9a5a35e8cd299457 : SUCCESS in 1h 48m 22s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/421d62f230034298969a0632e1918977 : SUCCESS in 1h 39m 27s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3685e7daba844bdb88711a33bf9ea40 : SUCCESS in 1h 55m 12s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a6a0fb4424149948c7f907c49b86a95 : SUCCESS in 2h 02m 16s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6465a09f2b2049f0bcbb36c9f7a2269a : FAILURE in 38m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de415939073c46f198e57c184439db9d : SUCCESS in 52m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/560b53ef18df4f5da1621978084bcdb3 : SUCCESS in 30m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e06169047e433d801e541ab5391a12 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e5e003a3bb2435cb24d4b2c6287d620 : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":1},{"id":"089b7a40a9d13ec62cfb40d3ba350d21f62c935d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-14 14:18:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92edd2c4769476a9ea5af98221d28b7566b30fbc","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-05-15 03:12:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e84e085b9176a10f6f40469985ba959d3d43d28c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:13:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2eae0129e37c890e3a56f8dc3b32bc6789ba526d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 05:45:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8432721e84f44674a0d3dbdfb080bfc2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f36eff47f4c4e9a916fbf6e8c4547ae : SUCCESS in 4m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c8c17b6e96a54cca9107438bd6983117 : SUCCESS in 40m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b84f3ca4f05a43c5a6dd6a9f88ff2053 : SUCCESS in 46m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51d7f9a6994c4813ba3efc1dd6b24f8a : SUCCESS in 45m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d12ba8187054060aad52023903e95c1 : SUCCESS in 45m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63a99a81f585461fb4fdf9d8e410ee7d : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dec07018edd42ae846c6489b5c433f0 : SUCCESS in 1h 09m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab00d3fffe1147f0a302a50eae471bf9 : SUCCESS in 1h 15m 02s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18bbf13409514ecfbced58d6774f6cab : FAILURE in 1h 55m 36s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e80aeca79d164d4388da57696bc165e0 : SUCCESS in 2h 06m 02s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1275780174c4641b6a10dc9865de73d : SUCCESS in 1h 56m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7af84ccffcf047938fd3c70cf8c09e14 : SUCCESS in 2h 31m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b6a346114f741819aed68aea72a6fb3 : SUCCESS in 5m 52s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3234821dec84b7aacc0d4a16ae49c71 : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b4970a725e275dd0ef692b881cb61b3859d16d61","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 07:21:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbfdbe274efcde02da3693c2a0387479cd5ee000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 09:51:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43571a4d9adb458b9316023a056cfeac\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f5c1a1000e1454e8c4169d4ea8416eb : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/eb1d023efff1411b9e36cd54e2089898 : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cbcc30eeb1f4e61926e18fc6950c881 : SUCCESS in 1h 18m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d71bc98f678478c96c9e28973ddbbc6 : SUCCESS in 43m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f4da2b3efd5473db02aff4cb88c1895 : SUCCESS in 48m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b844e3152f24fad912997c723c28f3d : SUCCESS in 1h 23m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/02c652327d3a402eb1a28c87b6e17534 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab8ed7320ddd4dce87efcfb4ff935f9f : SUCCESS in 45m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9def12732e884423b365763ce31e8a0f : SUCCESS in 39m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2316204798846e3bc50777bc845eca8 : SUCCESS in 2h 00m 32s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef3350fad119422a940a8dea61d2570b : SUCCESS in 2h 08m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2e61d4a36114eb19b2b547d56e41f1d : SUCCESS in 2h 05m 14s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6d8286dd0274bcc9f3a98876e23238d : SUCCESS in 1h 48m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3913d866d3784f569751a1c6c2b872f3 : SUCCESS in 1h 53m 03s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acbb95376f384aee85cff2d34e769512 : SUCCESS in 1h 48m 51s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/062dce1b7ce14ee2bb8119023405b27f : SUCCESS in 1h 57m 28s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f366d01ab92c4480a69141b5b18e03a3 : SUCCESS in 1h 58m 58s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73808bc3ba64405895f5076843b56a18 : SUCCESS in 2h 23m 31s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3e708b41849403b8b881c01b1685254 : SUCCESS in 2h 19m 39s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d8a1837aea0d46d8a02f4ab21889f2ec : FAILURE in 39m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e7c6d639b2634bd2b6aa509e80962dd1 : SUCCESS in 59m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a27d180901c418e82739dfd36b3f413 : SUCCESS in 58m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9eb521d3174fd892452664531e9793 : SUCCESS in 5m 41s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c080233d96047c8a028892f216e63e2 : SUCCESS in 1h 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"85aad068697849f9345cb3eb10cb3b364671c816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 09:53:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ae9345f91c64a3c030d183f938207d2e8dc9977","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:34:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2753c6b0d7b4ba7981d013930e8f82b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7caf2cca15114b42a860a8cbbaaadff9 : SUCCESS in 6m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/39ca2787dff8426f97c620323dc90bf2 : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a4ca93b45034e72a0c3cc0200079062 : SUCCESS in 1h 01m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ec49e7e91fb4dc48eef28ab8ea2d3ee : SUCCESS in 1h 10m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e485d038f0040819b02adb38bbfa839 : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5053ca1ae5f845a48104b3f952c631ac : SUCCESS in 1h 14m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a95dcdb5d6004a0d9e62d4fd7fb4baa6 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e233e828f44a482d9e868f84c6554af6 : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50d1b459427f4be6aea9245ea2d50964 : SUCCESS in 1h 56m 53s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c8150cbfdbd49919011e135c39234e6 : SUCCESS in 1h 49m 17s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5940cebb5bb641f79e2a5eb10c47110b : SUCCESS in 2h 17m 56s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c1e15198d454ac89e36e882663bd554 : SUCCESS in 2h 29m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad043a204024ba3b16935760d922f04 : SUCCESS in 3m 51s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5011f1c8f7a74f86bfaf3f873ee474df : SUCCESS in 1h 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b2bbc64092ac6553f30ccdc65e53326156e4ac84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:34:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9fc6d0824977b2047eab1e206334ab9e7388d829","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:35:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32e9d9f9f74e4f92af3168713dded828\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09454271ccc64dd3a0f4bd381efebb03 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ceb2d45abda69244a3e5bf1dbd78e3f7d403d2f6","revisions":{"ceb2d45abda69244a3e5bf1dbd78e3f7d403d2f6":{"kind":"REWORK","_number":1,"created":"2025-05-13 16:27:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/949653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/53/949653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/53/949653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/53/949653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/53/949653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/53/949653/1"}}},"commit":{"parents":[{"commit":"c5ad57d456967dd189cb130fa4315f3dd11a1a1f","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c5ad57d456967dd189cb130fa4315f3dd11a1a1f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-13 16:27:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-13 16:27:10.000000000","tz":120},"subject":"Use dynamic include instead of static imports for conditional tasks","message":"Use dynamic include instead of static imports for conditional tasks\n\nWhen tasks are needed conditionally it\u0027s better to include them\nto avoid all skipped actions inside of the them, which also consume\ntime.\n\nChange-Id: I0728383786d6f2282888fc1bf6ddace7773f2e40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ceb2d45abda69244a3e5bf1dbd78e3f7d403d2f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ceb2d45abda69244a3e5bf1dbd78e3f7d403d2f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
