)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~954269","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Id5991cd29051ef9638172a4a1ea168d6d12db21c","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-05-11 08:46:12.000000000","reason":"Change was abandoned"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-07-24 09:59:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-05-11 08:46:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id5991cd29051ef9638172a4a1ea168d6d12db21c","subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","status":"ABANDONED","created":"2025-07-07 15:29:54.000000000","updated":"2026-05-11 08:46:12.000000000","total_comment_count":12,"unresolved_comment_count":3,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"407cd826237cf9bb6c1c8826ef42858d9430c227","_number":954269,"virtual_id_number":954269,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-11 14:10:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"reviewer_updates":[{"updated":"2025-07-07 16:12:02.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2025-07-07 18:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-24 08:10:30.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"dbdee0c77c696d51d581d9ffdd7a0f0da4b11795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-07 15:29:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75b69f0857ac2448887ed6d6eee98c833b6f515d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-07 15:30:03.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"bc5ec4137fe0d060d32b37f8590e07be78562bfc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-07 15:51:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c470e65884d71bb37b7d6a13daf0d8d52e15e0f2","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-07-07 16:12:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85c902c2ea3107c05c474c86bb0b409fb37f6605","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-07-07 16:22:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76153ab33956b6bc10536f3097c70b88efcfd8d3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-07 16:54:31.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"4f732142068efd4741e3045539b731986bdae299","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-07 16:54:39.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"740495b5eda039e5dc87aa23cc4f7919e6fbb3e9","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-07-07 17:40:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f705c9c1ad9f8331b0a74156e80a31aa7d7299a7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-07 18:18:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be067eec63c7166c030dcee1eb054593457a68a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 18:47:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b63aa0bbf9f49ea8de220b32e9b1c88\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ac724f6082e4422baf271db745ba02f : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ddbc4a42c2894fabb9aa22367245ce17 : RETRY_LIMIT in 8m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b592411f7d0406f8f5898fc2b125e86 : FAILURE in 35m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20ca9527104945cdbc37065b2fb26525 : FAILURE in 39m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c875374f31741548323c8a66d6ebb56 : SUCCESS in 2h 08m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0416129a7a8d4c3ebb1ac5232c1fb45c : FAILURE in 35m 54s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/851229888e504f67b023e8d1a1e4bcdd : FAILURE in 35m 09s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dec77ddd81914f20898739abe486b204 : SUCCESS in 2h 47m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c647b49bf6945dfbfc66dd454d4f2e4 : FAILURE in 45m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3241ac90544049069a92b9b74f6bc720 : SUCCESS in 3m 25s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1233e906a7b40de9f689d5c1256550c : RETRY_LIMIT in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4a6e5d7dd3dcc200a6f05b7fcc8420060056bbd5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-08 08:38:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e03fd3ab5c8ca841c0119b8fdfc74956892f6ba4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-08 09:09:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9125e817203fff54ae80376057caa245c9bd8d55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-08 09:27:56.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3a18a48c5c2b69c8e9938e794261133622f0efc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 09:30:33.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/954239 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"b7a8f3489af8d9f88c790a4b390c286cb96f946a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-08 09:31:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"789b7875b297707452754f80f0dc859861ffe679","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-08 09:53:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bd92170b1e2c2d52210c9a229516abbb8d53e2d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-08 10:52:08.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"f77659919035d2872fa04854ade33036acd823ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 13:30:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3018b3c21ade48d793f13529902d8c30\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e3263289e67451fb457f2418befb44b : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/142b49894109464c8e0fcfce4ad753d5 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22b2ec38e780467cb54eb15157030f79 : SUCCESS in 1h 24m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca2a5b2619d847db9716167f0bffb02c : FAILURE in 39m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7cbfb1e63814a2a94f3f89f82958d69 : SUCCESS in 2h 04m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/230828848b1a44339080d3e1591ac256 : FAILURE in 1h 29m 51s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eb4661a5da14da2b6a203fedfd47643 : SUCCESS in 1h 56m 33s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92a4b632939d4b139acba7b3cbb7560e : SUCCESS in 2h 28m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c70650ab16a4399b0576846320a689e : FAILURE in 52m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a05930510043e288d2c520c70bb663 : SUCCESS in 5m 26s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eebc46123f9487e8a31084199c86833 : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":7},{"id":"f255df83ed21b64c35209fd4bec0394e62ec0f00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-08 14:40:50.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"660928a47c6729d90fdc7e6c906df6dc2433aab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 17:12:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd5c37098a564ea4b8d0508b5c7b2a8a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a22aa3c6a8641338577c99cc7374c8a : SUCCESS in 7m 53s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5121e109b49141b08cf2890f4d9b7192 : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db20c30fad4a4199a92b1cd40380a71b : SUCCESS in 1h 27m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3051c24be9c44609b54322601c8fdc73 : FAILURE in 1h 09m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb2ecebd54704c4f9437266de46e4e5e : SUCCESS in 2h 22m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac79eba4eb614bfdb78de3eff485e1b2 : FAILURE in 1h 38m 20s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19c67601327d49c99e2bca4e88bcc1ff : SUCCESS in 2h 08m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f08c012c6bd4b91aace89d98e700e38 : SUCCESS in 2h 17m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3e15c7838454fb087c9a188bc248337 : FAILURE in 58m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838ec5b513be4f28805b4f899a4ee111 : SUCCESS in 5m 28s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ec729ca6f9b49a0a40f14a487107084 : SUCCESS in 1h 31m 13s","accounts_in_message":[],"_revision_number":8},{"id":"009694a851f1908516eb0f0596446e7b8df40290","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-11 11:42:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"055dbc9c72c0aabe32710e9a4ecfc8d8d38214b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-11 11:43:09.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"47e9e3e5beb941e3216c42fd3e3ddd806413399b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 14:10:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e9b5d2f97e44d9a8eba1d8b27b0491\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35afdd86670745ce81dee93d3e7cb3f9 : SUCCESS in 8m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bb380c9be08c4d53aab0009e093b7211 : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6ba1d67729c41d6b1cfd1e9e9dc6287 : SUCCESS in 1h 25m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63da797479f24af1b2a8e9c4e53d2ab4 : SUCCESS in 1h 26m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/444a3d965bc140f690a89d3465338f87 : SUCCESS in 1h 59m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a42b1f22c8834203bcfe73e06da94bd4 : SUCCESS in 2h 19m 35s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/726b023e90f949b98ff98d674f1e168e : SUCCESS in 2h 17m 05s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4821a1e83f9e4e9abec4ba7e0414434d : SUCCESS in 2h 03m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dd6118971014aac88f18b9ab24394dd : FAILURE in 59m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a27868b3d341e0b23235b2608d51c3 : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c59a5ece1f04ab7b18c286d8e72b6a5 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":9},{"id":"a2a26432ff8fecf3dd7c7f44020fdd463600dde5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-24 08:10:30.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0e7572161c1f023787d29d8f9db8f703b8037e26","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-24 09:51:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bb88a022ec21e56ae41d46a984acf8d02f259a39","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-24 09:59:27.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5dd1ce3df1a84ce692242124c0fdc46796e3eba2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-24 10:32:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9451bf87a5d3ccb404549e79efd84cbe4da1de00","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-24 10:34:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"407cd826237cf9bb6c1c8826ef42858d9430c227","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-11 08:46:12.000000000","message":"Abandoned\n\nI think we have that implemented already","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"cb8bfc2655c5c4a73e0b893e703ddf45ddfc4de4","revisions":{"e55d1db1507f1911593c187b28ccf56b63a205b4":{"kind":"REWORK","_number":1,"created":"2025-07-07 15:29:54.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/1"}}},"commit":{"parents":[{"commit":"bb14d9c79f6265911f3dc7947ef257a41a3843de","subject":"Switch from wsgi script to wsgi module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bb14d9c79f6265911f3dc7947ef257a41a3843de"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:29:29.000000000","tz":60},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e55d1db1507f1911593c187b28ccf56b63a205b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e55d1db1507f1911593c187b28ccf56b63a205b4"}]},"branch":"refs/heads/master"},"f952ed0ebd6fcf2e3ebbb6f7a85d3ea79d8d163a":{"kind":"REWORK","_number":2,"created":"2025-07-07 15:51:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/2"}}},"commit":{"parents":[{"commit":"bb14d9c79f6265911f3dc7947ef257a41a3843de","subject":"Switch from wsgi script to wsgi module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bb14d9c79f6265911f3dc7947ef257a41a3843de"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:51:58.000000000","tz":0},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f952ed0ebd6fcf2e3ebbb6f7a85d3ea79d8d163a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f952ed0ebd6fcf2e3ebbb6f7a85d3ea79d8d163a"}]},"branch":"refs/heads/master"},"e0805fcbaeb2a46150d03a9e6011cf270037c59d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-07-08 08:38:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/3"}}},"commit":{"parents":[{"commit":"bb14d9c79f6265911f3dc7947ef257a41a3843de","subject":"Switch from wsgi script to wsgi module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/bb14d9c79f6265911f3dc7947ef257a41a3843de"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-08 08:38:05.000000000","tz":0},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/954316\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e0805fcbaeb2a46150d03a9e6011cf270037c59d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e0805fcbaeb2a46150d03a9e6011cf270037c59d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3af50431b5b20bc1d21c54d536c561a0c68d8d98":{"kind":"REWORK","_number":4,"created":"2025-07-08 09:09:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/4"}}},"commit":{"parents":[{"commit":"70a60c7941302b031d37955022af120d6ab3c8a7","subject":"Switch cinderstore job to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/70a60c7941302b031d37955022af120d6ab3c8a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-08 09:09:38.000000000","tz":60},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3af50431b5b20bc1d21c54d536c561a0c68d8d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/3af50431b5b20bc1d21c54d536c561a0c68d8d98"}]},"branch":"refs/heads/master"},"34bdb05657171d53305408bfb7e8eedc7457befe":{"kind":"REWORK","_number":5,"created":"2025-07-08 09:27:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/5"}}},"commit":{"parents":[{"commit":"70a60c7941302b031d37955022af120d6ab3c8a7","subject":"Switch cinderstore job to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/70a60c7941302b031d37955022af120d6ab3c8a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-08 09:27:56.000000000","tz":0},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/34bdb05657171d53305408bfb7e8eedc7457befe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/34bdb05657171d53305408bfb7e8eedc7457befe"}]},"branch":"refs/heads/master"},"5c76abc44fe38f60be2fe5d9120d59ce23472b90":{"kind":"REWORK","_number":6,"created":"2025-07-08 09:53:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/6"}}},"commit":{"parents":[{"commit":"c51c845231a83e69b9b31d5fec16bb71ab5740fd","subject":"Switch cinderstore job to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c51c845231a83e69b9b31d5fec16bb71ab5740fd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-08 09:52:56.000000000","tz":60},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5c76abc44fe38f60be2fe5d9120d59ce23472b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5c76abc44fe38f60be2fe5d9120d59ce23472b90"}]},"branch":"refs/heads/master"},"f88b8e16d10e44823d82814a48984de4a781977c":{"kind":"REWORK","_number":7,"created":"2025-07-08 10:52:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/7"}}},"commit":{"parents":[{"commit":"c51c845231a83e69b9b31d5fec16bb71ab5740fd","subject":"Switch cinderstore job to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c51c845231a83e69b9b31d5fec16bb71ab5740fd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-08 10:52:08.000000000","tz":0},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f88b8e16d10e44823d82814a48984de4a781977c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/f88b8e16d10e44823d82814a48984de4a781977c"}]},"branch":"refs/heads/master"},"6af8b13843293efd6259fa3a8dbdcdbcc3db3f65":{"kind":"REWORK","_number":8,"created":"2025-07-08 14:40:50.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/8"}}},"commit":{"parents":[{"commit":"c51c845231a83e69b9b31d5fec16bb71ab5740fd","subject":"Switch cinderstore job to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c51c845231a83e69b9b31d5fec16bb71ab5740fd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-08 14:40:50.000000000","tz":0},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6af8b13843293efd6259fa3a8dbdcdbcc3db3f65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/6af8b13843293efd6259fa3a8dbdcdbcc3db3f65"}]},"branch":"refs/heads/master"},"cb8bfc2655c5c4a73e0b893e703ddf45ddfc4de4":{"kind":"REWORK","_number":9,"created":"2025-07-11 11:43:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/69/954269/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/69/954269/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/69/954269/9"}}},"commit":{"parents":[{"commit":"c51c845231a83e69b9b31d5fec16bb71ab5740fd","subject":"Switch cinderstore job to noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c51c845231a83e69b9b31d5fec16bb71ab5740fd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-07 15:24:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-11 11:43:09.000000000","tz":0},"subject":"Use \u0027name\u0027 to specify SSL certificates to the PKI role","message":"Use \u0027name\u0027 to specify SSL certificates to the PKI role\n\nThis hides internal implementation details of the PKI role so that\nnew backends can be added easily, and gives a cleaner integration\nwith a user supplied certificate in the `src` field.\n\nA user supplied certificate where the src evaulates as \u0027truthy\u0027\nwill be installed from the absolute paths provided.\n\nCertificates specified by name will be retrieved from the directory\nstructure managed by the PKI role on the deployment host.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/954239\nChange-Id: Id5991cd29051ef9638172a4a1ea168d6d12db21c\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cb8bfc2655c5c4a73e0b893e703ddf45ddfc4de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/cb8bfc2655c5c4a73e0b893e703ddf45ddfc4de4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
