)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~974624","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~Ib62934e38b199a066949ffb5e7804ac7a5dea9f6","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-25 12:01:13.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-02-25 12:01:13.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-02-25 12:01:13.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-02-09 12:15:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-25 07:01:33.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-02-03 19:29:57.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ib62934e38b199a066949ffb5e7804ac7a5dea9f6","subject":"Remove deprecated scrub_time","status":"MERGED","created":"2026-01-24 15:52:50.000000000","updated":"2026-02-25 12:02:23.000000000","submitted":"2026-02-25 12:01:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974624","meta_rev_id":"82eb55a3f35c09af189eb2bf6c7b34fed2246ab8","_number":974624,"virtual_id_number":974624,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-25 12:01:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-02-09 12:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-03 19:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-02-09 12:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-24 18:28:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-25 15:00:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-01-25 15:00:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-02-03 20:29:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-02-03 20:29:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-02-08 19:31:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"95fe1a3a275a9bcd8b2f6c997e7c335af345c129","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 15:52:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1621642ea102e5de81af5e449e1c65f67995f487","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 15:53:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"95cdb6fb2c96035af3baa92673acc73c8fea0584","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 15:57:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4ad98efe137e8417267a52ccd7d19e4a6ab69b45","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 15:58:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"053b3e9053e0a2ea9c40076d626186e7926923cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 15:58:54.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9065090bba4f89e78bf4433f8206ebaf90df2769","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 15:58:59.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"a936a46059583356c07408614677fecff6796dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 15:59:48.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"a6b71686985168468d3c585589fb1565a1a52ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 18:28:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0cde2a0754a42798b7d4fc6227d9f99\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/882a9d1f315a4f718081628653508326 : SUCCESS in 11m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e0014139072148eb95c7ea18610e720f : SUCCESS in 48m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/21e73a0fdd9d435fbfdfc8c7cc09fc7d : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/4a164b56e21d481fa5c5273568c0f67e : SUCCESS in 48m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d42e381c5d8e485186a7a2adbc8668b5 : FAILURE in 1h 33m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f33c26bdc2c74667bb354cdd62889ae6 : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f532de0963174920813340326e195d50 : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/720259d8a90b46a8ba7df2606fe5607d : SUCCESS in 1h 26m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/337b58774e0a48819e8c2d144f5ae2a8 : SUCCESS in 1h 43m 00s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/59c24e9c00524eab84a27d565bd27958 : SUCCESS in 1h 54m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e6e76e0e993429bb2fc67518efbac80 : SUCCESS in 2h 18m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c04eb36a7a14bdfb33fc78ca02b6957 : SUCCESS in 1h 45m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/decf737ebabe42ecb81e521db5e62e4c : SUCCESS in 2h 01m 38s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b3bf8a367af499a92544bb7c8fc20ef : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60bf39bd67bb4eb2a007a7b7b3e2bb12 : SUCCESS in 2h 26m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5105c4d1b7e34f01938b135be5efd423 : FAILURE in 12m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3e8cb1be8c04ac298697d7003a1ef39 : FAILURE in 44m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e328bcfcf1411d8d8134b277076fb6 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86782b0ef25f4f0b8720686c1c20213a : SUCCESS in 2m 36s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a64aced79ba4ed184822fd4e409854c : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":6},{"id":"d574eb84f7302f40f3b37ba62638a5d08b804f4e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 20:30:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"71e5f0f4a4ee4b3c5a76649eefba349c97d74bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 23:27:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df1d2f24dd594693a59d55cbd259a469\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f49bfb4760642409ff1d46b76694813 : SUCCESS in 16m 33s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6a1461eb4643484fa1546c49742bdb09 : SUCCESS in 1h 34m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/74c68f6aa5a24debbf71f173241d198c : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/14df6c12a1234efcacf6c0b1faae5c99 : SUCCESS in 1h 27m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f65cf50e1f5c43559aa26e8812332ba0 : SUCCESS in 58m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b148742e906d4ec2949b1ee79c646eff : SUCCESS in 1h 18m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33af6dbbf3bb4364b2e14167b1f42dda : SUCCESS in 1h 39m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c597222066034d998bad1f529b563c2e : SUCCESS in 1h 40m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d78b2b7290614d98988d25b31491b39d : SUCCESS in 1h 42m 02s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/638326b58dc94b939192154a4e7e8211 : SUCCESS in 1h 14m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/304b579e93c64465bacbbdb6942fccd0 : SUCCESS in 2h 09m 41s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4655e51139249d289329c2d3657b9e6 : SUCCESS in 2h 03m 50s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2836fcb3222e42ea848bd96a41315a81 : SUCCESS in 2h 19m 09s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4935ab772cb6411a9eef4c7ac41c0e2f : SUCCESS in 2h 32m 55s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f077149f649452a89dd8a1a743ccc3f : SUCCESS in 2h 47m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8dc1b3c2e7904fb6b48100c90cac67d3 : FAILURE in 18m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f309bedcf3da46c58b971c5e0aa11182 : FAILURE in 1h 00m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0422c93e5f4afb84058a854d9ea3f3 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7747b34096cd426f878f5e7c47aa8073 : SUCCESS in 4m 54s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0236584bcaf4c0a8d5e24217dae53eb : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":6},{"id":"9898f9c43241bbb899f694f90b519c0f724c86e5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-02-03 19:29:57.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ff8741c4679e543b1537c90b419c35bd1dfb9c8e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-09 12:15:46.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c32a7ad6faf12e449788d1ce680eb988cc39c638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 12:17:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9700c56e3a6ff41bd8f67f131bb68ac7655f848f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 14:28:08.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/242f2472a1bf4f07a4b5fd3e20c7c69e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4533cc4b37a449b9ac857d184a94c9a0 : SUCCESS in 10m 29s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/820aba42f8a44c7ab611e294b7e994a4 : FAILURE in 36m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ccc0d9d127c14261bdf31e484259f219 : FAILURE in 1h 09m 03s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/1f971b89661542e5b7c6ffeb98336e06 : FAILURE in 1h 09m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/925315bbf20a49d9b17079a08d83fcd2 : FAILURE in 40m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f37565e6273b4216b3ac3ddb50eef029 : FAILURE in 1h 00m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c6157196ab54b7d99d5ccc8b0704200 : FAILURE in 1h 08m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaed988c4430454b9e87f5a12740ae68 : FAILURE in 1h 27m 56s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52858ae293c941f6a89b5f61db6ad757 : FAILURE in 1h 45m 36s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0421362cb0448568ec6626344c12b1a : FAILURE in 2h 05m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb88ac4abe04830b47b5c38e544384d : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5e41418d8ed4376bb60214c93585f73 : SUCCESS in 2m 27s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f5b5027c6be4eee8d6d7377ebb38bdb : FAILURE in 1h 15m 13s","accounts_in_message":[],"_revision_number":6},{"id":"01302a4ac863c4b6991a21f7f01d36c08ea5a3be","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 11:15:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b5d54232237608db0e5691103ad0b6ad8d4deea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 13:57:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a19a2f8dae764d5cb0d25022851a80ce\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af4e90c99f6245e9b211087cb1c7a4e0 : SUCCESS in 8m 58s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/fae698e7ef784c5a98e0af26c8a48241 : FAILURE in 39m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c15c77d5c9f34d46b889686412dcc895 : FAILURE in 39m 55s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/516c4aff6d574f9c8e02286cee347fed : FAILURE in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/691e0ba4356640af92c875295d5c1b62 : FAILURE in 1h 19m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/63d72ad904344f11b3d35dd922d86bbd : FAILURE in 40m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4085a040a74c436d9b31ddcd84f86054 : FAILURE in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/217edf2804b744cc867a22b72de55c9a : FAILURE in 1h 26m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/23fe0e9c58e9462fa15ed7cdd8e4770c : SUCCESS in 1h 52m 19s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0fa50da8b78a4efeb2eef71cf9f8073f : SUCCESS in 1h 00m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de98ac74f39c4e73b228420a8155c09c : SUCCESS in 2h 10m 30s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73a9781a2c0242978717f1540c2aaf7a : SUCCESS in 2h 11m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f88ae67ff764163bee9f78c718e7606 : SUCCESS in 2h 25m 55s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7f95ff6c6e746a6a71f5594a9a49c7c : FAILURE in 1h 22m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f177428269de432f82c8a7468571ba14 : FAILURE in 1h 41m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0e30f88c432f40efa27495649ddf2676 : FAILURE in 10m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e37f556b94a47c8add5988dba3a0faa : FAILURE in 1h 01m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ae9e120aa94afbb3c6762ffd9e4225 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10f54b0f976d458a9efc9b656d85d1f0 : SUCCESS in 2m 23s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4622d93fcd340b8bada6e7678db7afe : FAILURE in 1h 09m 29s","accounts_in_message":[],"_revision_number":6},{"id":"8c8565b761c34d389952245126f68bce115f611e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-25 07:01:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5fe41b99dbebc96fc1054bbbf7f39e101095a369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 09:23:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2276ae4cf3a34d9d9641ddea677311c7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7d11bd429364f57998315fe2b1fe259 : SUCCESS in 7m 36s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7863320a3cdc4d86bc42c5b6afce8410 : SUCCESS in 41m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ca01c92c293c4dd58a07be9534f5565d : SUCCESS in 1h 00m 50s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7aceb5380c6047bea8571c7d9f6149a6 : SUCCESS in 1h 16m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8125e51374ae46e4b29d7a451198f308 : SUCCESS in 1h 05m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fccec2cf9b9049e89c1ba80977a96692 : SUCCESS in 37m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64bb16c625494b84946744aa49d63e97 : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01c4e65e20524a23aba975044f0a66fc : SUCCESS in 1h 27m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/904fcac4851a41bc8374d66756ef57b5 : SUCCESS in 1h 30m 57s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d4f765ca4d56499c8f4db8ffbfb44f6d : SUCCESS in 2h 14m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc679d9ff22d42c998ea61d0f515c373 : SUCCESS in 1h 01m 49s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea2dd294f1294d248c2863bc0b0ada7e : SUCCESS in 1h 35m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/292d614079674ffdb84b53c16186a6a4 : SUCCESS in 2h 11m 35s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07b69e8832e343f9bf264a31f9c68126 : SUCCESS in 57m 57s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/401da79160424fafac7bc4ed481b20ca : SUCCESS in 1h 36m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c489daae289344dfafbd0c4baf6542f6 : FAILURE in 8m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83be64296c53422fa370a55a8fdedd40 : FAILURE in 44m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d8dc1fb04314f5ca135c7017491f43b : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfe47fe35db944cdb34601039a485273 : SUCCESS in 1m 46s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/141f4b34a00d45e9a1bf369e3a895421 : SUCCESS in 1h 19m 30s","accounts_in_message":[],"_revision_number":6},{"id":"568dec5ca98bad62793a243c570782a91472af79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 09:25:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b56a8594187cea578698506928f6fc6157295d13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:01:12.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69589b40cb984937884321b5bdd96ae4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c85d1e993324d4c96dc0681321b6f15 : SUCCESS in 7m 16s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3585205f507a48d78783245eb5a0ab17 : SUCCESS in 36m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7173a5e21915445497da4ef76b5b8536 : SUCCESS in 36m 37s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9697c0b734a048ea8afef1a345b0af2a : SUCCESS in 40m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d32fed1f6aa64270ae2e63e1d6aba3a0 : SUCCESS in 39m 05s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2782e85fb01a49f895e5dfd0c154b64c : SUCCESS in 39m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/259480e18b584719ae2ab19e26efe1ea : SUCCESS in 34m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7508b57341b4925bed98973f1328bb7 : SUCCESS in 1h 20m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0afe8fcf5b004ee1bd518a1a897765ef : SUCCESS in 2h 01m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26bba9d91f914950874d14033c969b78 : SUCCESS in 2h 31m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b26954255b46bdb69dd55b62ab32d7 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e77f36bfa5b443158ad348b6dbc7a021 : SUCCESS in 3m 53s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32cf3d6794f649d9a78bf912a15c6ba8 : SUCCESS in 1h 15m 20s","accounts_in_message":[],"_revision_number":6},{"id":"02e40ac899b06b9c72c2e9945995d5c8bbf8aebc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:01:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"82eb55a3f35c09af189eb2bf6c7b34fed2246ab8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:02:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b0447de7079492c990333a5adfabb66\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1914f549f43a4bfe9503c1c91b80bd6f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a106eaf1cfe954d015cd7b57d97ea9ad8afe76d6","revisions":{"049e04364a89e59565d859a51d282dc82386c868":{"kind":"REWORK","_number":1,"created":"2026-01-24 15:52:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/974624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/974624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/1"}}},"commit":{"parents":[{"commit":"4f3a17a5e65065eb16d4af364094fffa5189d233","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4f3a17a5e65065eb16d4af364094fffa5189d233"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:52:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:52:50.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ib62934e38b199a066949ffb5e7804ac7a5dea9f6\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/049e04364a89e59565d859a51d282dc82386c868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/049e04364a89e59565d859a51d282dc82386c868"}]},"branch":"refs/heads/master"},"deaa4a30803d8363cb372cfea088d36fb00cd2e7":{"kind":"REWORK","_number":2,"created":"2026-01-24 15:53:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/974624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/974624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/2"}}},"commit":{"parents":[{"commit":"4f3a17a5e65065eb16d4af364094fffa5189d233","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4f3a17a5e65065eb16d4af364094fffa5189d233"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:52:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:53:28.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ib62934e38b199a066949ffb5e7804ac7a5dea9f6\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/deaa4a30803d8363cb372cfea088d36fb00cd2e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/deaa4a30803d8363cb372cfea088d36fb00cd2e7"}]},"branch":"refs/heads/master"},"ba86633ed15a8e2ceb44783e29ebec5ff29de0ba":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-24 15:57:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/974624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/974624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/3"}}},"commit":{"parents":[{"commit":"4f3a17a5e65065eb16d4af364094fffa5189d233","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4f3a17a5e65065eb16d4af364094fffa5189d233"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:52:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:57:47.000000000","tz":0},"subject":"api: Remove deprecated scrub_time","message":"api: Remove deprecated scrub_time\n\nThis parameter was deprecated during the past cycle because it is not\nused by glance-api actually[1], and has had no effect since then.\n\n[1] bd6fb8f1756b004aa66af389810cd7200603c2d3\n\nChange-Id: Ib62934e38b199a066949ffb5e7804ac7a5dea9f6\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ba86633ed15a8e2ceb44783e29ebec5ff29de0ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ba86633ed15a8e2ceb44783e29ebec5ff29de0ba"}]},"branch":"refs/heads/master","description":"Edit commit message"},"39e650543981dbc690db72adf3d17ad758ca5de7":{"kind":"REWORK","_number":4,"created":"2026-01-24 15:58:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/974624/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/974624/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/4"}}},"commit":{"parents":[{"commit":"4f3a17a5e65065eb16d4af364094fffa5189d233","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4f3a17a5e65065eb16d4af364094fffa5189d233"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:52:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:58:46.000000000","tz":0},"subject":"api: Remove deprecated scrub_time","message":"api: Remove deprecated scrub_time\n\nThis parameter was deprecated during the past cycle because it is not\nused by glance-api actually[1], and has had no effect since then.\n\n[1] bd6fb8f1756b004aa66af389810cd7200603c2d3\n\nChange-Id: Ib62934e38b199a066949ffb5e7804ac7a5dea9f6\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/39e650543981dbc690db72adf3d17ad758ca5de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/39e650543981dbc690db72adf3d17ad758ca5de7"}]},"branch":"refs/heads/master"},"835017ce58909ba321449f4c0c67195d859ec65c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-24 15:58:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/974624/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/974624/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/5"}}},"commit":{"parents":[{"commit":"4f3a17a5e65065eb16d4af364094fffa5189d233","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4f3a17a5e65065eb16d4af364094fffa5189d233"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:52:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:58:54.000000000","tz":0},"subject":"Remove deprecated scrub_time","message":"Remove deprecated scrub_time\n\nThis parameter was deprecated during the past cycle because it is not\nused by glance-api actually[1], and has had no effect since then.\n\n[1] bd6fb8f1756b004aa66af389810cd7200603c2d3\n\nChange-Id: Ib62934e38b199a066949ffb5e7804ac7a5dea9f6\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/835017ce58909ba321449f4c0c67195d859ec65c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/835017ce58909ba321449f4c0c67195d859ec65c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a106eaf1cfe954d015cd7b57d97ea9ad8afe76d6":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-01-24 15:59:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/974624/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/24/974624/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/24/974624/6"}}},"commit":{"parents":[{"commit":"4f3a17a5e65065eb16d4af364094fffa5189d233","subject":"tox: remove linters and functional from envlist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/4f3a17a5e65065eb16d4af364094fffa5189d233"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:52:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 15:59:48.000000000","tz":0},"subject":"Remove deprecated scrub_time","message":"Remove deprecated scrub_time\n\nChange-Id: Ib62934e38b199a066949ffb5e7804ac7a5dea9f6\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a106eaf1cfe954d015cd7b57d97ea9ad8afe76d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/a106eaf1cfe954d015cd7b57d97ea9ad8afe76d6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
