)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~976064","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I78fabeafeb4fd9b98386670334c17c06f79db866","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-03-09 13:20:48.000000000","reason":"Change was submitted"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-03-09 10:42:24.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-03-09 13:20:48.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-09 13:20:48.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-09 13:20:48.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-03-08 07:04:04.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I78fabeafeb4fd9b98386670334c17c06f79db866","subject":"Remove deprecated digest_algorithm","status":"MERGED","created":"2026-02-08 19:28:50.000000000","updated":"2026-03-09 13:22:25.000000000","submitted":"2026-03-09 13:20:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976064","meta_rev_id":"a552ef00377e1e96df26c113fd3b646bb4b2eed7","_number":976064,"virtual_id_number":976064,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 13:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-03-09 10:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-08 07:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-03-09 10:42:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-08 19:31:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-08 19:31:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-02-08 19:31:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-02-08 19:31:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-02-08 22:07:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-25 12:22:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"b8a7c45a1eaa532395dc3a6fbfc125cbcf0c0d7b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 19:28:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"297b4cdabc7f6f4a6ff48372be6141111b9b3d49","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 19:31:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4070fb4e4a9c8a41abfd2b0e0a41ba2823dd8732","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 19:31:15.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"4ca15f1da27c2f4fb31d9a208b35476e04dd169c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-08 22:07:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed5af721846344a28e4c67c54b43a83a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a23b1ec09286420fac8e639446c7fb3b : SUCCESS in 9m 11s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/79e2f1b9a14846d9b82164465b4d3a83 : FAILURE in 1h 09m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/82d4c8f7fd9043ceaa611d620f89c18c : FAILURE in 1h 16m 31s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ef55540fdd024a1ca5e60c32b212b87e : FAILURE in 39m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f90951e48a5f4f1ab2d99523467c6f56 : FAILURE in 40m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b8c8da3cc98748768bbbb488e3a8f6ad : FAILURE in 1h 09m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39a5bb2b9ad44ff8833d72de9ee60ed9 : FAILURE in 57m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/331743f535394829b09219c51361b2e7 : FAILURE in 1h 09m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2b7359c736304684acc38896e2ff6fd9 : SUCCESS in 2h 18m 23s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/426d404eb2a64c1b92eaa316fff348b7 : SUCCESS in 2h 31m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fe12df5698240939d6d7419016cf7de : SUCCESS in 1h 06m 07s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b40a9ab890e4b54baed87f3c3cead07 : SUCCESS in 1h 28m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5e5a73f81144909ab979c5ee2c873d3 : SUCCESS in 1h 45m 23s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4da09e23ff654a749c89cc1ccb009bd0 : FAILURE in 1h 53m 57s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f277c0ab1641be95406136b7a50ba5 : FAILURE in 1h 10m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b03790fe09f2424bae4171ba06c87c08 : FAILURE in 7m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/735fb76849f84662af4d8ec9e2a96cc4 : FAILURE in 54m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc64f6c9b8dd4530be9e7d706ba0e0a4 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/399e7de9e6404ba3b76c1b8d11368bf9 : SUCCESS in 2m 41s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92375c0e23834f93a0f64f4ffa6fa0be : FAILURE in 1h 16m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f09cbf67d96c0ed9077a0343f6193bb84f6a5c5d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-24 20:14:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d38c6bff7488f480fb5498902c2414101b6840fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 22:40:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/081cb6e7e0bf4b56944b5cdb578e084e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a4289a7b58b4f86b0edd23bf74f9e7a : SUCCESS in 10m 27s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6f34d2214e2e42cab3659d95f89cdde9 : SUCCESS in 43m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1621fe426cee4717a4db8f69743ebc74 : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9e69b9b7e7de4e32bd5f0667c7a4b587 : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2b95af595bd5489a8eee21f81a0d0837 : SUCCESS in 44m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ccde6897151a4ee7988b6f6dec8d5941 : FAILURE in 41m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55ddcbad339c43bf93fe9af2cad6bb3a : SUCCESS in 1h 16m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f75680639d374790b7da53ce0477a5b2 : SUCCESS in 41m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1c339e6b8d204017a369fbe09959c5b5 : SUCCESS in 1h 44m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/81278c1267d44b82a633effd02677d9e : SUCCESS in 1h 01m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68f7e1936f3743e491d2641c3c8fde91 : SUCCESS in 1h 00m 58s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe204839cd3b4fdeb283249185b9e19f : SUCCESS in 2h 19m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/829c8006d9fb4e57b3a801d5c7b547ec : SUCCESS in 1h 01m 12s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40e72ec77b93405abcbbdd063b102bff : SUCCESS in 1h 08m 14s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b24bb0be48014479b15893e6a56ad846 : SUCCESS in 1h 54m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bcf6b67e3d714c8eab84ca645e8b7120 : FAILURE in 6m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d415f34931e943c3810c65c2285c0acb : FAILURE in 24m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41007e8935f5492e8b11b412be0d61db : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf00858559c41c9947895f048457529 : SUCCESS in 1m 40s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b15f8088254a442dbace7280792869f6 : SUCCESS in 1h 23m 21s","accounts_in_message":[],"_revision_number":2},{"id":"fc35526a75c66d0a893d82e1ad6c8553ca65e857","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-25 07:01:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5332b862eb3a66b558ec576a8d934dc488071768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 09:29:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/408fe1d68f5e4cffb754c18089907948\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/766cdf5bc0eb4d15862aa88a9afc2ef1 : SUCCESS in 9m 17s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/703a78feee5d48cea48166e3185ad418 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/848e016e2f45441b9c225ebad3badb89 : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bffe60681653461494fbce34cfaa90f2 : SUCCESS in 1h 09m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3bfa5b19b8bc4771b40f70fae031e16f : SUCCESS in 38m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ea884fe70e04e5db46852a951cebc72 : SUCCESS in 1h 24m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8bf7085917b40fa88d48e8c2063ccfe : SUCCESS in 41m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4ea39519ae143b5b8c229c8aafe5162 : SUCCESS in 1h 20m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0f7b25b283e04c759d7edbc9be2c960d : SUCCESS in 1h 32m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dc77b56dc4384acc8c9661de34cc5973 : SUCCESS in 2h 15m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea46e9a2f0804da3b538fc8c42fdc9b0 : SUCCESS in 2h 19m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/107e2356ba2c4887887ef10a103a4032 : SUCCESS in 2h 18m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/673f0c1fa62e440eb6c44a1f9efa76ec : SUCCESS in 1h 00m 47s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/831a9f85fc4845edbe1bfe48909ac9fe : SUCCESS in 55m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04f462a4aa7347ec99a54e7f7b4d4a10 : SUCCESS in 2h 18m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/17159e236c2e48d8978d5ea3155dc6cd : FAILURE in 8m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7755f58d940b457e98507ffa5f65d9ee : FAILURE in 42m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/258717a73f5e4824ae1260715b8b9cb6 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aaad5ab3ba44fc48ed8a349da1d8eb4 : SUCCESS in 2m 42s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1be4a44ee634b7881dc8d4f59956b8a : SUCCESS in 44m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f0d347e650667aca8ca2099382e621c8022f4276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-25 12:22:28.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ef3729946e5b942b27ecf151baa0366fceb9fcdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 14:47:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ff09feee3d429088b5e4b8049f2769\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d16c4e3612f2454daba2c7c8a3ecde29 : SUCCESS in 10m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a082bde812bd47dca01b2640ea29e2f7 : SUCCESS in 1h 18m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ea74b876d0794e519e794d856af16f71 : SUCCESS in 1h 25m 42s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/76b4c96a52c2431496114fb510f61958 : SUCCESS in 42m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2ab22e01f697498cb2c651040fbf8d17 : SUCCESS in 39m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/28ab62f4c9ad40abb1beea51332168fc : SUCCESS in 1h 06m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44c2717f0fb44f4daf72e331354974d7 : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f25ca8beee740bcabbc0ac4002a246e : SUCCESS in 1h 21m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4172fad74edb452eaadba792e9c76a1b : SUCCESS in 58m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6409b60deec648f6bcc0d79f290e8a7f : SUCCESS in 1h 02m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c31cf0ba2a724b109fddb46427e03440 : SUCCESS in 2h 18m 39s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18990808a18b4c9c9f101b0b556e6e62 : SUCCESS in 1h 39m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41f9f369e29e4b39be4da9e56438348b : SUCCESS in 2h 08m 05s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c6a29d065da4730a6aa3460365d5002 : SUCCESS in 2h 06m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88e57883c8d0480b9c3a8c7624b8d22b : SUCCESS in 1h 47m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e8c73c8aac94471a8b1d0ebff06734dc : FAILURE in 11m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e485a8d59f3b4b7bbc866edef3f04d41 : FAILURE in 55m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/511142ca528c4c95a8ba91b3b5c36c3a : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79405652649a46659f6f6ddcbc71e055 : SUCCESS in 2m 37s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9db9822ee401401face42df631d05824 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":3},{"id":"8695d655c20046d3c676db08e56a09a6827e6250","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-07 11:09:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95e7fee5c065e0417af5a5d5c61add05ceda8e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-07 15:02:30.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3d3d5d244d90975131d311988a62c7fec1c3a07d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-07 15:02:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9d2460488ed76ff044bba10b652ffbe68948891b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 17:27:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be08e3bc09d4447da719eacc2d9306c9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/520ec01f0b534d27bf531f448922ffef : SUCCESS in 12m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e0cc651f729e4f8985891aefb64e63e4 : FAILURE in 1h 02m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c1393c54d7e34c0fb83dbc7c464d3056 : SUCCESS in 1h 07m 50s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/49552e4fb5784f87a47bdcba9f5976dd : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae20ce43e11b462d908211f9ab6cfc77 : SUCCESS in 47m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/16cf17d33f6e4823b558581e24450b11 : SUCCESS in 46m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb20c3ccf8e24e708b1e10fd0332f269 : SUCCESS in 1h 28m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b6a5486f79245c691079ed7d458c029 : SUCCESS in 1h 31m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fc6b5538c888485c9b67144de38b8583 : SUCCESS in 1h 00m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/57a3dee8fc23428cba2a80a297f2bb16 : SUCCESS in 1h 35m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15f29e8dd7564e128dd5987f3e7be55c : SUCCESS in 2h 15m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bd0e5633b7643aba8d5cfee8ffb9f54 : SUCCESS in 2h 15m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89e9fcf6c51f488ea71f4864fa266589 : SUCCESS in 1h 01m 46s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3083eea3a50244eb96e3b04179675c59 : SUCCESS in 2h 13m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c34f8fbcdf24d86857f986ae4e6b700 : SUCCESS in 1h 19m 05s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/10b386c2bb70463b90ede4a967fe6d12 : FAILURE in 7m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f3ddd499ef44ce3a43d6ac5a2c32606 : FAILURE in 26m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c09401af9149c88aa3f49c8571cef9 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fabe5bec6dd4c2d8000f62d2daf889e : SUCCESS in 2m 26s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75561179f1364eecbe55de952586766c : SUCCESS in 46m 08s","accounts_in_message":[],"_revision_number":4},{"id":"cc0eb96af5fafb355e37a7c9d2e46a4bad2d92be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-07 19:24:15.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5f0a2f8e0b3d7b3ecf7a171f0adfe469acca7766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-07 19:24:26.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d404ab43ddce386b18c07521b45b7933611f3dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 22:07:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df69fea0caf84363adb6098a0b12be43\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cde8221128246cdbb595eb3fc4a0725 : SUCCESS in 14m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d18cd3d3701b473e8ae0cfb90e6e59fe : SUCCESS in 1h 08m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c5d1747a7da7430983a955b088bc92a0 : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7ddd2925102c495b82d4d0929fbff1c1 : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/65d802c1385a4220ab7c472fe379b189 : SUCCESS in 1h 39m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/538dde7085e0455192e2735a99ee2f8b : SUCCESS in 1h 38m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/513a5e8d08db4805a8fa6bf6073d826a : SUCCESS in 1h 30m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b50feaedf0042ccbc0103cd80c4ae25 : SUCCESS in 1h 34m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/981983e3dc5a45939887423034ec34f0 : SUCCESS in 1h 47m 30s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a95793f000cc4725b4d89d8f9deef0b1 : SUCCESS in 1h 34m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e216041b25ca43b8abb04c4af1f51910 : SUCCESS in 2h 10m 24s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0602fc741d95423aafb9551a623b82a7 : SUCCESS in 2h 20m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50ed0fad4c7a4ffb92f94f6dccf40f97 : SUCCESS in 2h 23m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79994b89f4524f1f8b4493401c7a5cf2 : SUCCESS in 2h 13m 09s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/109317945cfe4e6a978caf649f8d8e69 : SUCCESS in 2h 36m 12s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e00c1eb18814be19565da0ce94ccbfb : FAILURE in 7m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00d0693e7ef643dfbb8053cb9a2bda9e : FAILURE in 29m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bcf9e176ef74f12951a9a97dade752a : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79fc95daaa3445c9b3da6c38aee6c4bc : SUCCESS in 2m 36s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/613c7446af1b4d6fabde9ceefc9dca5d : SUCCESS in 1h 33m 35s","accounts_in_message":[],"_revision_number":6},{"id":"a9630b71b4af01d1ae38a870f2d3c136ef537c36","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-08 07:04:04.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"580356aab1e0bd6fcfe9d6415c524ca26fd39106","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-03-09 10:42:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4349279547d5fc43dafd42f926127352dfb9a011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 10:44:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"81958be69491a0d606c10fd3dac998536e4e1c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:20:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7681ed4f2a54865a15aa10e65c7619b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/419b0e07c7284f7e83fa8b45a188f0a9 : SUCCESS in 12m 39s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/438732766135451e909a7a1484edf4f5 : SUCCESS in 48m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/11513be0d1a54b9da624eb24ced7c0a4 : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/75350204210441aab86f1937e68513df : SUCCESS in 1h 32m 45s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/70789dc5c4734c2794629d5f68ac24b3 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/78a6239869194f95b440a4b321ded060 : SUCCESS in 1h 39m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c5678fdda464ec9afbf3c776a626269 : SUCCESS in 1h 30m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f58499474e34572a7f0c4853232cd3d : SUCCESS in 1h 32m 23s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d2ed70ede86477590b7f9380cd956ee : SUCCESS in 2h 03m 15s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf516dab01394072be4e2996f932dd78 : SUCCESS in 2h 25m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cbdec7affe84c38a16ec4b6037bda1d : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8ac515cd5184312a63f2445510cdf0a : SUCCESS in 3m 23s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1ff70940f1546c3a91a6d10c98c4b60 : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":6},{"id":"24810e1286a6259257afe93ae685deb0583effea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:20:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"a552ef00377e1e96df26c113fd3b646bb4b2eed7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:22:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c74484ff9d498cbc36554c65d5b219\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96de23a7354241e28d1b0bef3d7d0abe : SUCCESS in 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"349eea35a78241d6e257ceb217e9a1e274807d8e","revisions":{"151c7bdd13c56ce8c15ff29ca0e533e6056f2e18":{"kind":"REWORK","_number":1,"created":"2026-02-08 19:28:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/976064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/64/976064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/1"}}},"commit":{"parents":[{"commit":"ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f","subject":"Update configuration for cinder_catalog_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:28:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:28:50.000000000","tz":0},"subject":"Remove deprecated digest_algorithm","message":"Remove deprecated digest_algorithm\n\nRemoved in Glance: I7ad94bd2be45a25c9c1dccc552fa9110fc4862aa\n\nChange-Id: I78fabeafeb4fd9b98386670334c17c06f79db866\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/151c7bdd13c56ce8c15ff29ca0e533e6056f2e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/151c7bdd13c56ce8c15ff29ca0e533e6056f2e18"}]},"branch":"refs/heads/master"},"e033b3ca7769bc092ab37681d6616b3709843ef5":{"kind":"REWORK","_number":2,"created":"2026-02-08 19:31:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/976064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/64/976064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/2"}}},"commit":{"parents":[{"commit":"ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f","subject":"Update configuration for cinder_catalog_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:28:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:31:03.000000000","tz":0},"subject":"Remove deprecated digest_algorithm","message":"Remove deprecated digest_algorithm\n\nRemoved in Glance: I7ad94bd2be45a25c9c1dccc552fa9110fc4862aa\n\nChange-Id: I78fabeafeb4fd9b98386670334c17c06f79db866\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e033b3ca7769bc092ab37681d6616b3709843ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/e033b3ca7769bc092ab37681d6616b3709843ef5"}]},"branch":"refs/heads/master"},"24f98735a3f10b200532e3643cf993662df13834":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-25 12:22:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/976064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/64/976064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/3"}}},"commit":{"parents":[{"commit":"ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f","subject":"Update configuration for cinder_catalog_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:28:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-25 12:22:28.000000000","tz":0},"subject":"Remove deprecated digest_algorithm","message":"Remove deprecated digest_algorithm\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/913720\nChange-Id: I78fabeafeb4fd9b98386670334c17c06f79db866\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/24f98735a3f10b200532e3643cf993662df13834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/24f98735a3f10b200532e3643cf993662df13834"}]},"branch":"refs/heads/master","description":"Edit commit message"},"92dbaa64909046bc55ef38f9241c316b4b126326":{"kind":"REWORK","_number":4,"created":"2026-03-07 15:02:30.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/64/976064/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/64/976064/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/4"}}},"commit":{"parents":[{"commit":"ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f","subject":"Update configuration for cinder_catalog_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:28:50.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-03-07 15:02:15.000000000","tz":0},"subject":"Remove deprecated digest_algorithm","message":"Remove deprecated digest_algorithm\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/913720\nChange-Id: I78fabeafeb4fd9b98386670334c17c06f79db866\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/92dbaa64909046bc55ef38f9241c316b4b126326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/92dbaa64909046bc55ef38f9241c316b4b126326"}]},"branch":"refs/heads/master"},"366fa69e165bd7d101283b7cdcfc4514703696ea":{"kind":"REWORK","_number":5,"created":"2026-03-07 19:24:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/976064/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/64/976064/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/5"}}},"commit":{"parents":[{"commit":"ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f","subject":"Update configuration for cinder_catalog_info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/ffe7c1eed9b1d0a5d3ef9fe1bac8a1508430618f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:28:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-07 19:24:15.000000000","tz":0},"subject":"Remove deprecated digest_algorithm","message":"Remove deprecated digest_algorithm\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/913720\nChange-Id: I78fabeafeb4fd9b98386670334c17c06f79db866\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/366fa69e165bd7d101283b7cdcfc4514703696ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/366fa69e165bd7d101283b7cdcfc4514703696ea"}]},"branch":"refs/heads/master"},"349eea35a78241d6e257ceb217e9a1e274807d8e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-03-07 19:24:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/64/976064/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/64/976064/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/64/976064/6"}}},"commit":{"parents":[{"commit":"5fa4003d6a546c6788a66189ebb74c90eab6a664","subject":"Merge \"Remove deprecated scrub_time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/5fa4003d6a546c6788a66189ebb74c90eab6a664"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-08 19:28:50.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-07 19:24:26.000000000","tz":0},"subject":"Remove deprecated digest_algorithm","message":"Remove deprecated digest_algorithm\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/913720\nChange-Id: I78fabeafeb4fd9b98386670334c17c06f79db866\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/349eea35a78241d6e257ceb217e9a1e274807d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/349eea35a78241d6e257ceb217e9a1e274807d8e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
