)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~985580","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I5d2ddf445329b93dd1c30a3f0ab2fe7006604ce6","project":"openstack/openstack-ansible-os_glance","branch":"master","topic":"openstack/openstack-ansible-sync/sync-files","hashtags":[],"change_id":"I5d2ddf445329b93dd1c30a3f0ab2fe7006604ce6","subject":"Updated from OpenStack Ansible Integrated repository","status":"MERGED","created":"2026-04-20 23:15:11.000000000","updated":"2026-04-22 13:30:30.000000000","submitted":"2026-04-22 13:29:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985580-openstack/openstack-ansible-sync/sync-files","meta_rev_id":"4075fb3cdded9e8522253e9983053e98acf8ad23","_number":985580,"virtual_id_number":985580,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 13:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-21 08:14:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-21 08:14:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 03:11:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 08:14:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"9113704aafb442e0e0e49338f4a0fd0c7c8acd7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-20 23:15:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70ff6a377141ff36a525efad32c2eb5099e2265d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 03:11:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2c9a57735b4481a8a9b0161fbdab8f7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3262e132832d49f7888bbde28b44cf3f : SUCCESS in 13m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7191edeee4df4798bd93f3af6609223c : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e85b934a9f0746ac8976777c52f38eaa : SUCCESS in 1h 23m 25s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9569dac7bb3d43568b74f72ea363d199 : SUCCESS in 1h 30m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/80a551fae5dd474996c20bd01e69226f : SUCCESS in 1h 16m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/11934164705248e1b176e2b53cb06907 : SUCCESS in 56m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5ea95ecaede46968622f9a55df57845 : SUCCESS in 1h 26m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50b8aa2192084b8db6f8e226b8d2be1d : SUCCESS in 1h 44m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9aad5d604e8e40b1a1b20c45df48f4f9 : SUCCESS in 2h 25m 20s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/93afe6134ea3457d9bc9514a444c9fcf : SUCCESS in 1h 14m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c01541ad7c064a459d97e9d9089cbb02 : SUCCESS in 2h 16m 56s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5651b1488ec43fd8327f1233fba5e1c : SUCCESS in 2h 24m 16s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/567448f167384956b79ea5c08599d1dd : SUCCESS in 2h 14m 08s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8859b2c990c04c7cb4c88400d7a20f3f : SUCCESS in 2h 42m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1127664df245439cab3a0ec6e23b81ff : SUCCESS in 2h 32m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/51c03fb3cdc449278b0842aaea0c65ce : FAILURE in 12m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8ea7e06de6a4a3f897341eb056351f8 : FAILURE in 59m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa65f73515a54772970f8430aa119e09 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b6ae0ff9d24524a8b093a3f1775728 : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/815a22e8811b4d4c85da7fca374c8b9a : SUCCESS in 1h 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3f9a22e45aad662c2956e4fff5f973ce6b5131ed","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 08:14:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6dea020db3a8194768e790ea4a7826adf3a12383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:18:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"008f90bf2a9cb3737e9ad2ee63f82e8dec13a695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 11:52:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06676f301d7a4c93bc8829850045912e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f6ed2391d9a4167b4cec0bfd5a8dd6b : SUCCESS in 13m 23s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f04fff5db6c48daae7b2e0532e7ef97 : SUCCESS in 58m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2f5de7a7b78b43b096204a31256cc222 : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a2b5eff2f50c454abc01c7240f217cfb : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/afd2d5d5968245668c2aba739ccc513b : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9ed73ac57d334ad29ce10f93e60ae0de : SUCCESS in 55m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a268147e2b964e58b9cfee23b2cf741b : SUCCESS in 1h 36m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c001d3164778411d86a120a6ba440692 : SUCCESS in 1h 47m 25s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51b2e580135b42028f8ea008cfd00496 : TIMED_OUT in 3h 11m 20s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c50ace434e744f669acdd11022fc7a86 : SUCCESS in 2h 56m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028df9cc03e14700a764d322e3e3cf42 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d099d0264901461abbf6fed817ff2dae : SUCCESS in 3m 52s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b279324c6d94e2f9242967ce5b5958e : SUCCESS in 1h 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9b34cba301a2d3d262cad050cce2b2f6947bd796","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 12:23:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b30f7d3dbd89ce35f1c1eb417e3f79624cdef79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:46:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/621d8f4b368048708239d6724eb5ffe1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d399cfde42042f7a173253dac60fdfe : SUCCESS in 14m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/467bd1557eeb4469a5658da3108d3129 : SUCCESS in 58m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2c40dca52dc44a32b29b5a7dc27d9a13 : SUCCESS in 1h 30m 06s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7523edf3290e4396aadbadf5c30fb50b : SUCCESS in 1h 05m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f46d33fa012947e0b73aa238b8c280f1 : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6e4fe7991d9f4d50bf7ca33c34bd67e5 : SUCCESS in 59m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dd85ae2cccb46d5a54c3551b4764dc7 : TIMED_OUT in 3h 12m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a362a545fd0b4cafbce2bdd1dcd7cd1c : SUCCESS in 1h 34m 06s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2ea113e3fd334432b636ce8d1973a704 : SUCCESS in 2h 24m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/52faf5a8d5d34281b86ad9b47dcf3eaf : SUCCESS in 2h 07m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fee0befeeb50499d8bacf5f051b3d13e : SUCCESS in 2h 28m 11s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7048edf0fba74147bd74146de5008ffa : SUCCESS in 2h 18m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75876aeb30ab43cc967f2f6769320973 : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec9f9723756c444797e8c34aacafe771 : SUCCESS in 2h 53m 56s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9815402db5184c099a826c4bae845ef7 : SUCCESS in 2h 49m 55s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ff69ec1554014ff682242ed68e509cf0 : FAILURE in 9m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ad849eab48a457ba7ba0f858744f4db : FAILURE in 50m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0108aef7dfca4c04843f2b3e2ea22316 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/723b99adbb4f41dcad01de751208e2dd : SUCCESS in 3m 02s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f0b13e5df0483cbf3b8c87632ba2ea : SUCCESS in 1h 41m 33s","accounts_in_message":[],"_revision_number":1},{"id":"97ea1ecd8458775511e85e51761337375d0eafd0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-22 08:04:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e2a81c362d0234556ea50e7db49dcc1f4713076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 10:53:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/200a1219a74a4c978aa348f988c884cc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/562dc24912834d4cb9bb8c874d313afb : SUCCESS in 13m 00s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/87b37bc5201f4d1d8ee3a6e860b45bf2 : SUCCESS in 1h 35m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0bac4a6c52934f40a7cfa15275069c80 : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/67d5ec584ab040e09e838266af2b591e : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e1cb554cb54a4c09b394d52364c8609e : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9487f3acceb140c6bb1dcd5fe82ca805 : SUCCESS in 1h 21m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c91f0a485e14485c9657d3ed4ef37b8a : SUCCESS in 1h 28m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55b0c740fa8e4cd18407c288561f8080 : SUCCESS in 1h 37m 29s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/863f86b4231b429a9eff681d348b2d6f : SUCCESS in 2h 25m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b64cbbc3c58a47c8a7e9e895f420ba92 : SUCCESS in 1h 24m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e6d9376b9984c47b6e4b9f1ece85222 : SUCCESS in 2h 34m 13s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddc0d2d0306b4d989eb64530cfc24950 : SUCCESS in 2h 15m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66db1d0eca10405082714c345cc1acc5 : SUCCESS in 1h 10m 57s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c29acc60f5a4f9ab17bd5e1f4571b27 : SUCCESS in 2h 33m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d7ce437686f4a2bb54f12687f8bb221 : SUCCESS in 1h 38m 26s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/185f8265674942c4bfc12ac495f38489 : FAILURE in 11m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dff39c4aa84c4f9e8a55ecd0d650624a : FAILURE in 32m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b28d69bebea94e46be637c022f6cd41d : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7533a9b699a40ddb9b95b3b292d6101 : SUCCESS in 3m 45s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a1d7533aa4d4f5dabb7009b4a164be9 : SUCCESS in 1h 18m 06s","accounts_in_message":[],"_revision_number":1},{"id":"899ecee421809321c1506a07dc8697dec85c1267","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 10:55:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70574d8662db1df6c49f68f3d6446016f4cc7df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 13:29:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ed638fc11b5415297348f945fcd0bb0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dca887c01a94c0f910238f135e509d7 : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/191c61c274754645b45f63e45897ec61 : SUCCESS in 1h 15m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/98c55c8cde8c4a6abbff99369d51f533 : SUCCESS in 48m 04s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a3d18e455a9c40dab0f3632167fe572a : SUCCESS in 1h 23m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/886a2599116c4ba9ab648dea16fbbdbf : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0e9399cfcb9749fab056026f3e71cce2 : SUCCESS in 49m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c67d662a3724fe287a7688d2b3e27ad : SUCCESS in 1h 07m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/632d8d859ff14efeadad7bf4a9e90fad : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0306aa86aa724b3eb5814f84086c65f5 : SUCCESS in 1h 11m 43s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a60f3ae1feb841fe96b408776ba33851 : SUCCESS in 2h 32m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2406121ea9a4b4c8251e27515171ae8 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09e0097039a744f69df61002d74a02b6 : SUCCESS in 2m 21s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58f05edb133e4e0bb37185d82d6d5f25 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5ef731c9a7c5e4827003d365eb7a62606fdf9a5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 13:29:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4075fb3cdded9e8522253e9983053e98acf8ad23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 13:30:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97bed6f7a68941ea9f9c46c1cf14cf9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c32251954f47a88e4d526b65c7b74d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c48d1b1deb1c74877569d255f1de5d46b57e5c48","revisions":{"c48d1b1deb1c74877569d255f1de5d46b57e5c48":{"kind":"REWORK","_number":1,"created":"2026-04-20 23:15:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/985580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/80/985580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/985580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/985580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/985580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/80/985580/1"}}},"commit":{"parents":[{"commit":"0191966704ab0a9cacc3e56897fca303dc6115eb","subject":"Require uwsgi to run api service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/0191966704ab0a9cacc3e56897fca303dc6115eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:15:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:15:10.000000000","tz":0},"subject":"Updated from OpenStack Ansible Integrated repository","message":"Updated from OpenStack Ansible Integrated repository\n\nChange-Id: I5d2ddf445329b93dd1c30a3f0ab2fe7006604ce6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c48d1b1deb1c74877569d255f1de5d46b57e5c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c48d1b1deb1c74877569d255f1de5d46b57e5c48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
