)]}'
{"id":"openstack%2Fopenstack-ansible-os_glance~986285","triplet_id":"openstack%2Fopenstack-ansible-os_glance~master~I34668859a6cac1acfe6924657e1ece73284937a3","project":"openstack/openstack-ansible-os_glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-06 16:12:56.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-05-04 13:09:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-05-06 16:12:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34668859a6cac1acfe6924657e1ece73284937a3","subject":"Sync schema-image.json","status":"MERGED","created":"2026-04-26 03:07:46.000000000","updated":"2026-05-06 16:14:02.000000000","submitted":"2026-05-06 16:12:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986285","meta_rev_id":"725e83fc9d11c18ff9ef25041d31b1b95d33bb02","_number":986285,"virtual_id_number":986285,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-06 16:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-30 19:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-30 18:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-05-04 13:09:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-26 06:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 17:26:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-30 17:26:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-04-30 17:26:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-30 19:05:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"1d7901ac5ca9afaa7cbc776f9663823090279bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-26 03:07:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddffc305c8b085fe39ea5bfa5b366e85b2adda5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-26 03:08:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dd7e57636163060e209a70774fc413d5642f4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 06:12:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b2a51be29c54915a28121565c260223\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/865670b9a13844eb983d03050b178cfd : SUCCESS in 13m 52s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8948bb9107874a7aa60b19867655a532 : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2b5ced5fede94164b34de7312f02c060 : SUCCESS in 47m 32s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e69976d604a54b7792bd36820a70340f : SUCCESS in 47m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d1f62569b623448d9e51e40c0420d65a : SUCCESS in 1h 11m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/61b5a0beeebf47a089e7f976e578b4fe : SUCCESS in 1h 38m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c32819fc0b0e44a9b4c877e56ec55fbd : SUCCESS in 48m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51114a27501741e6a3b4b9c829232f23 : SUCCESS in 48m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/72313f263592428fb805fc6938cff696 : SUCCESS in 1h 42m 53s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/17b29d73476a413eae76a16951013743 : SUCCESS in 1h 52m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/789970f157794164ba90c695c53d73c0 : FAILURE in 2h 09m 34s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76becb9e80f34705bd47cbac7e2659c2 : SUCCESS in 2h 15m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed2023ab46d84d04aba7a1d493660143 : SUCCESS in 1h 07m 05s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b8c041e2ce54208b39b7b3c2f877c00 : SUCCESS in 2h 22m 10s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa573c45b651401fa94197d877cf7f8a : SUCCESS in 2h 55m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c1c388e2655f414a8b2a30bbb2b9a35b : FAILURE in 12m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f390cf0cf066499ba56ecb5211d15b72 : FAILURE in 59m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2017b3e7624d0490f1f40e2be07e78 : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd4ede07bf854ec2b9d1c8842e89d51d : SUCCESS in 1h 30m 49s","accounts_in_message":[],"_revision_number":2},{"id":"38dc0e71581b55ab0861a56779973d70e7cea558","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-26 13:50:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bf109d96a0c96036754dd99b66871265585d10c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 16:52:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55fc91889e4c4cb3973c8e747c26149b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cc9c184008a4178930f7b6d339a9012 : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/87a13b3edb6e43209e81a8c7dab8e99a : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a098ec3ff1de41d4b9b0d61cd9bc1c1c : SUCCESS in 1h 26m 20s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0305cc3146644c4eb6746cf08faeb464 : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1da22842df884e108d0749b0b667b167 : SUCCESS in 50m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/db67524c74444c5397581e5b278d635e : SUCCESS in 52m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba819889b8d14b639fd7a83b9c1f42af : SUCCESS in 46m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0f9ba7adaef440db8df8d1938529519 : SUCCESS in 1h 35m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f4ef12e199424581ba1221b7bcc13b65 : SUCCESS in 2h 57m 32s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aa91ea179f4640068cc7774d7eac4527 : SUCCESS in 1h 08m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dec5d18aaca243e581208eb1d969c58c : SUCCESS in 2h 25m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa8db77f56754f01ba5932e6de759fd7 : SUCCESS in 1h 39m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32afcd6170e745d1bfb9d2fdd79eeb13 : SUCCESS in 2h 12m 59s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/150095bf1a4349e3984f04ebdfef84a8 : SUCCESS in 1h 42m 47s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/327b5afa080549e7918a29b9a3dc4e6b : SUCCESS in 1h 49m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a7f72406e0f54ac4a67f6c526a26ba83 : FAILURE in 9m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95c91fbd62df477fada1fed9c13b9506 : FAILURE in 43m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dbb06c76a44ebfbddffc1a87ba3f89 : SUCCESS in 6m 54s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/138c1d389e284e67a3a349d0dec1e7e2 : SUCCESS in 1h 28m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d4cf4a6d5c5af46cbe038551a59a0f3df4cba516","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-30 18:50:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e12be1dcf1f4bbe8139a36c918df599dbc6f2f74","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-30 19:05:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c227c7e3e43b43d6a985dcf54d1ef3222e9968d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-04 13:09:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"184353a73d9886ce6e3338e9c2642fb17de83356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 13:12:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9396bee8d6ff32e71d779742355bc6518c3203b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 16:21:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0ee0fd1380b4cce9d0e7aa4e7d69f6d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/359351e4017d444bbe7704a3619fa09c : SUCCESS in 12m 58s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/47ac0d09cbb7458eb96fefc86817ef44 : SUCCESS in 50m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/caa9e31d70b34cc18293bc42eba79400 : SUCCESS in 1h 31m 56s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/112ecdfb7fed4f05939c6c843011d45e : SUCCESS in 52m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6419a35d40a343dab7f7e7ef4006cf6b : SUCCESS in 1h 02m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/25db108f33f94c2fa0f7b52629f49bc9 : SUCCESS in 52m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f6267efd9f54fee857addd75af3fba7 : SUCCESS in 1h 32m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db86cb484bff46c895e9e92bb53a84b4 : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85148c9942294d15a3bda805544132e5 : TIMED_OUT in 3h 06m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c39e31c7b34549178a5232186e9297a8 : TIMED_OUT in 3h 08m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57dc09dfb24488296cc64ec24108313 : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f70e9941356478492e218398d5f693a : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":2},{"id":"665bdfbf850870265dcf6e6ce5f6652de41b0f30","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 16:22:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b09fc6ea0992404c242a748bb89d7874e628a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 19:33:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/859b2713c98946629a8ebef60183708e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2682d339ab0246ed9dd32eeaf90cbe59 : SUCCESS in 11m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0be06408215d46deb76bdf377697603c : SUCCESS in 1h 31m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/82221ae7bd474b69b4630c6d35842eab : SUCCESS in 1h 32m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/611c593915a8432782c1c1923cde09b5 : SUCCESS in 50m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/be8dc2f04eea4458b39f8cfd1e4a9d18 : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/901006e7cf1444b1b352581e0c763bce : SUCCESS in 1h 27m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c68f5abb44ee4cb29ede5bd590efa12b : SUCCESS in 50m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fedafda305842bba102985edc9f63d5 : SUCCESS in 1h 27m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7ddc0fa519444c47962fb18919755b1c : SUCCESS in 1h 42m 32s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/339ada4e9e2046548771be2d6f9ba2b7 : SUCCESS in 1h 09m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aabfa4087b8d42b8ab1bf2ac712b4e59 : SUCCESS in 1h 58m 48s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb1f379d1a2446c296c20e077ec56835 : SUCCESS in 1h 53m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdeec70151bf4ce385dc3e113649bdc7 : SUCCESS in 1h 41m 48s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16a2cf832af447cab0d56da4682a2909 : TIMED_OUT in 3h 07m 20s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3486090d36294e1789872954fd9255b3 : TIMED_OUT in 3h 04m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4459d57d35be460a89b5a5fcb7bc2c2b : FAILURE in 12m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f955cf9e036416fb62a1bcaeaa737c9 : FAILURE in 58m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736c7cafa647498a85b3d5d6dbfc62ba : SUCCESS in 4m 07s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a9946e8e0db44ba97a31278a0adcc45 : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"defc6a57ffac34d3b9944179bb1f11a0261ac2ea","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 08:21:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c673d72d74204adabbf9c01064163358a2d29cc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:06:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/825539712ce94db282590c2f38398af8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c2778391f2f4220b8e3fca8ac1d90d6 : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2bcd8442b93d4c05995af050f802f504 : SUCCESS in 1h 03m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/aa1c599378144211bbaa3463f290a763 : SUCCESS in 1h 00m 26s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/2f4fd9e3a7f54fe2b3d9b60f95a4bc37 : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/73ab6b494dfc44b6882507870b11dc42 : SUCCESS in 1h 35m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6674f5779b2d49b08a115910314a6c59 : SUCCESS in 49m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a52bae0ce9847a28f18012aaf6fcad4 : SUCCESS in 1h 30m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/369ac8b4b1af43de81749048c5cb638a : SUCCESS in 1h 30m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2344bf23f3834ab982d066bc39aabbb3 : SUCCESS in 1h 46m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/25f18d5e45f7414e8fce35e24cc27295 : SUCCESS in 1h 53m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0220ebf616d24911a29a5ac1f7d41bb7 : SUCCESS in 2h 32m 08s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1adbb957765b4c1cbc3af52b117f9c22 : SUCCESS in 2h 15m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2f03bfe722040b9856c090ca49201d1 : SUCCESS in 2h 27m 10s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e907631d846f4506af6296230fe98255 : POST_FAILURE in 2h 25m 44s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfd822fe1ab845cf87bdb1cecf669ce7 : SUCCESS in 2h 37m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b576832fdeb140479c5eff76dc78dc7d : FAILURE in 8m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718ca043d3674079a39b0967cdbb2ad2 : FAILURE in 1h 04m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92cbd278d6b04501a4e563c7fe9aca4f : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbaf1751ed3840b4a2c5580fce7f1e02 : SUCCESS in 1h 28m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2d9a61fba15fc6821998c28dc44ff794ef10ff9a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 11:07:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fe67dae0fbe61fbdc6aace01ae017311ab769f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:34:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a44068eb9c984e17a96ca72113741139\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cde553c841a41d38b74ddf86185e75d : SUCCESS in 11m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2b80f69852b84494b39253f407036ccc : SUCCESS in 1h 13m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c1df7efb0e184a9fb3d722f23502c6c4 : SUCCESS in 47m 03s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/eff6c2a9f26048e3b1c22fcee42296bc : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9bb1d13ab18e44b1b1d4320f2630899d : SUCCESS in 51m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6d441af5c8af4b0b87b02fbba874bab1 : SUCCESS in 52m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aafc4c381f0b4610bd4f687fcfaf9506 : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e77e3730d43d40d8a87a797f20e5e895 : SUCCESS in 1h 28m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0fd4042550694cdeaad74ee07290432a : SUCCESS in 1h 32m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5f567ec2ad584081be7544249e5abbec : SUCCESS in 1h 29m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d40b60c46acf45ee9a60f580e666bef3 : SUCCESS in 1h 28m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/620a800cde864429bea60ace70a60faf : SUCCESS in 2h 10m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdb2ee9228b74b2683d7c5576941ffb8 : SUCCESS in 1h 34m 43s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aed53795a66423cb654394477fd7592 : SUCCESS in 1h 36m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43d7b1c2aae64810a3f8d1a6b501b290 : SUCCESS in 2h 19m 36s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/69059d85f2bc40de8ef0548227eddff4 : FAILURE in 10m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3bd01bf01c5477ebdec647094f6511a : FAILURE in 38m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc17267cc4a04a11a6ea6f681325f526 : SUCCESS in 5m 32s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/781f2435377243fab5e83aef809d97ce : SUCCESS in 1h 24m 25s","accounts_in_message":[],"_revision_number":2},{"id":"23637ed7bbcbd2e67a80cc32783747e6bd58a3dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:37:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f38cd020bffbee66ec0e4bd87dd4431c00cbe8df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 16:12:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d07416db504dc4beb3121f05a9b3ae\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cee6abf0015b44cf8a5b33bfc32c59fd : SUCCESS in 11m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f56221d5ba86496599b75ba16e073f81 : SUCCESS in 1h 32m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/388387f31e224647abda7bf0ba00a25a : SUCCESS in 1h 07m 35s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f0d5a229f43a44daac34c1be3c14c358 : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f05222c6bb1c4b96a039d9b317441623 : SUCCESS in 54m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/68c874fb9ded4f0a91dfe18b0092924a : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/804c633ecee040aab7877b5a2bdd557d : SUCCESS in 1h 29m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ebfcf21cbc34380ab43e0d52681e82f : SUCCESS in 1h 26m 53s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8f300b4cd434ed3ba88f66943337408 : SUCCESS in 1h 50m 16s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f31eac591548ba9aa7984bba08b1fc : SUCCESS in 2h 30m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179f51a8ef2e4d579e99b7ffa127264c : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_metal_cinderstore-ubuntu-noble https://zuul.opendev.org/t/openstack/build/924595628b454c768f606ab5199a390f : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b6587403c2a740d6092d772c5a3fd2f78c1a2bac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 16:12:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"725e83fc9d11c18ff9ef25041d31b1b95d33bb02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 16:14:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b5ed09eaf574da99edfa8ae8593f495\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328c049750aa4adcb6907a37856e7f96 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b78cb0c1bb3c96f1c1cf6e4822b7342dfc0605df","revisions":{"7bd13073e05e4ca2ac92c2255ef3fd93f59db250":{"kind":"REWORK","_number":1,"created":"2026-04-26 03:07:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/986285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/85/986285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/1"}}},"commit":{"parents":[{"commit":"c48d1b1deb1c74877569d255f1de5d46b57e5c48","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c48d1b1deb1c74877569d255f1de5d46b57e5c48"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-26 03:07:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-26 03:07:20.000000000","tz":540},"subject":"Sync schema-image.json","message":"Sync schema-image.json\n\nThe file is outdated and does not include a few items which is now\npresent in the file maintained in the glance repository.\n\nChange-Id: I34668859a6cac1acfe6924657e1ece73284937a3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7bd13073e05e4ca2ac92c2255ef3fd93f59db250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/7bd13073e05e4ca2ac92c2255ef3fd93f59db250"}]},"branch":"refs/heads/master"},"b78cb0c1bb3c96f1c1cf6e4822b7342dfc0605df":{"kind":"REWORK","_number":2,"created":"2026-04-26 03:08:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/986285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_glance","ref":"refs/changes/85/986285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_glance refs/changes/85/986285/2"}}},"commit":{"parents":[{"commit":"c48d1b1deb1c74877569d255f1de5d46b57e5c48","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/c48d1b1deb1c74877569d255f1de5d46b57e5c48"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-26 03:07:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-26 03:08:44.000000000","tz":540},"subject":"Sync schema-image.json","message":"Sync schema-image.json\n\nThe file is outdated and does not include a few items which is now\npresent in the file maintained in the glance repository.\n\nChange-Id: I34668859a6cac1acfe6924657e1ece73284937a3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b78cb0c1bb3c96f1c1cf6e4822b7342dfc0605df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_glance/commit/b78cb0c1bb3c96f1c1cf6e4822b7342dfc0605df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
