)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~614103","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Ia219e620c24d77d85d308db22888c0d46ee3f617","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","hashtags":[],"change_id":"Ia219e620c24d77d85d308db22888c0d46ee3f617","subject":"missing document start \"---\"","status":"MERGED","created":"2018-10-30 07:38:58.000000000","updated":"2019-04-24 20:26:09.000000000","submitted":"2019-04-24 20:26:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"614103-1556137569820-40cc56dc","meta_rev_id":"cf074d43c83a4764e74269a1ef1e6e1f0b33c934","_number":614103,"virtual_id_number":614103,"owner":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 20:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-04-24 18:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-04-24 18:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 18:33:31.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 20:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32ea62633c60b34beec5f1c206619b23c4777fa6","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-10-30 07:38:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f87b68e59147530ff58b3a8f7c093fa34469b6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 08:17:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/614103/1/check/openstack-ansible-linters/9b30e23/ : SUCCESS in 4m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-centos-7/62950a7/ : SUCCESS in 28m 39s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-opensuse-150/1300714/ : SUCCESS in 32m 31s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-opensuse-423/449b8fe/ : SUCCESS in 27m 41s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-ubuntu-bionic/083b614/ : SUCCESS in 28m 49s\n- openstack-tox-docs http://logs.openstack.org/03/614103/1/check/openstack-tox-docs/ff9f6b1/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"21c8c4a848430f55e15785e474dca3cbef98ad44","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-11-08 03:55:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"5c914697443cbe81799c99404dd4dbdd6ea02734","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-12-10 23:05:47.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"00dab69f0cf5069c335eb80a113d7fee699144c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 00:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/614103/1/check/openstack-ansible-linters/16d6db0/ : SUCCESS in 6m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-centos-7/d15e565/ : SUCCESS in 31m 29s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-opensuse-150/ca9ecf0/ : SUCCESS in 35m 44s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-opensuse-423/20b2b50/ : SUCCESS in 30m 27s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/03/614103/1/check/openstack-ansible-functional-ubuntu-bionic/fccc82b/ : SUCCESS in 26m 46s\n- openstack-tox-docs http://logs.openstack.org/03/614103/1/check/openstack-tox-docs/56c0a5c/html/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"02e854ee1358e77c55884757931b01de62079779","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:33:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd850132473fecb0b108faf7f5cf1534e4856c6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:33:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4f38cf47df0dbd504112b473da781fe3322b7c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:26:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf074d43c83a4764e74269a1ef1e6e1f0b33c934","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:26:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/614103/1/gate/openstack-ansible-linters/98b7e5e/ : SUCCESS in 8m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/03/614103/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/f17d446/ : SUCCESS in 1h 49m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/03/614103/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/d7b9a27/ : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/03/614103/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/bb6c7ef/ : SUCCESS in 1h 26m 17s\n- openstack-tox-docs http://logs.openstack.org/03/614103/1/gate/openstack-tox-docs/a1649c2/html/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1414ea661f96b71d0211266ac955a0ebeed9784e","revisions":{"1414ea661f96b71d0211266ac955a0ebeed9784e":{"kind":"REWORK","_number":1,"created":"2018-10-30 07:38:58.000000000","uploader":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"ref":"refs/changes/03/614103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/03/614103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/03/614103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/03/614103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/03/614103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/03/614103/1"}}},"commit":{"parents":[{"commit":"a50476d6de74e6089985701e30f013b1b198de22","subject":"Set container_name as host_var in test inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a50476d6de74e6089985701e30f013b1b198de22"}]}],"author":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:38:42.000000000","tz":0},"committer":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:38:56.000000000","tz":0},"subject":"missing document start \"---\"","message":"missing document start \"---\"\n\nChange-Id: Ia219e620c24d77d85d308db22888c0d46ee3f617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1414ea661f96b71d0211266ac955a0ebeed9784e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1414ea661f96b71d0211266ac955a0ebeed9784e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
