)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~654358","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I38d4526af4aca650fae2c438ffb07e51006792ec","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I38d4526af4aca650fae2c438ffb07e51006792ec","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:14:15.000000000","updated":"2019-04-22 12:48:46.000000000","submitted":"2019-04-22 12:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654358-1555937326708-7a5aba1f","meta_rev_id":"00a7e6a9de693ef51c3386c88e9074626022a1e8","_number":654358,"virtual_id_number":654358,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 12:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:52:20.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:52:20.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:52:20.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 12:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"580131bf319a15016ce8f18f60bdf32463a43591","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:14:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74117ea856f857ee32aab5d4ed16215dd95512c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:59:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/58/654358/1/check/openstack-ansible-linters/008afe3/ : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/58/654358/1/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/f483752/ : SUCCESS in 1h 32m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/58/654358/1/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/02474ff/ : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/58/654358/1/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/40a448c/ : SUCCESS in 1h 03m 24s\n- openstack-tox-docs http://logs.openstack.org/58/654358/1/check/openstack-tox-docs/354154b/html/ : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":1},{"id":"13af707293a8c0f266af873aadcded1907778359","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:52:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a9856d448b59c787c8e2e4458027ac5a8e152f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:52:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aff7d30cb110cf010acda32e9580de31893e37d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:48:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"10eac98d56bc9dd1b325d4cde0e343485e1ac43e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:48:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/58/654358/1/gate/openstack-ansible-linters/aa7236f/ : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/58/654358/1/gate/openstack-ansible-deploy-aio_telemetry_metal-centos-7/ff53139/ : SUCCESS in 1h 34m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/58/654358/1/gate/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/bad6b46/ : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/58/654358/1/gate/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/e5858f5/ : SUCCESS in 1h 07m 19s\n- openstack-tox-docs http://logs.openstack.org/58/654358/1/gate/openstack-tox-docs/e0cea36/html/ : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd7c6f412c52da0c6f6a259396e9f83b48f19a83","revisions":{"fd7c6f412c52da0c6f6a259396e9f83b48f19a83":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:14:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/654358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/58/654358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/58/654358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/58/654358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/58/654358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/58/654358/1"}}},"commit":{"parents":[{"commit":"3852789416f4e3ad84f84ad63aa751eb36bccf3c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3852789416f4e3ad84f84ad63aa751eb36bccf3c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:14:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:14:13.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I38d4526af4aca650fae2c438ffb07e51006792ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/fd7c6f412c52da0c6f6a259396e9f83b48f19a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/fd7c6f412c52da0c6f6a259396e9f83b48f19a83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
