)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~655544","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I2c952b52c2a812df1b7db96abbeb26092a6db80d","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"bug/1826250","hashtags":[],"change_id":"I2c952b52c2a812df1b7db96abbeb26092a6db80d","subject":"Adding gnocchi_service_timeout","status":"ABANDONED","created":"2019-04-24 19:58:14.000000000","updated":"2019-07-30 19:46:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df85d94724d996cd22aecb404bc04bb159e59bf4","_number":655544,"virtual_id_number":655544,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"date":"2019-04-26 16:36:39.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-04-25 01:06:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":-1,"date":"2019-04-26 17:34:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 01:06:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-26 16:36:39.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-26 17:34:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"41b8ed48cc96b6dccaf00e54172b9c6d0647181d","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 19:58:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48dbb10c8b0663a64d19863d549a54d2e4e099ba","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 20:12:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10e9c42fbdb4eea99c3d62b8bdae7598da3dea7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 01:06:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/44/655544/2/check/openstack-ansible-linters/f2fc96f/ : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 http://logs.openstack.org/44/655544/2/check/openstack-ansible-deploy-aio_telemetry_metal-centos-7/579c05d/ : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-stable http://logs.openstack.org/44/655544/2/check/openstack-ansible-deploy-aio_telemetry_metal-debian-stable/5086805/ : SUCCESS in 1h 08m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic http://logs.openstack.org/44/655544/2/check/openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic/e45d1cc/ : SUCCESS in 1h 06m 52s\n- openstack-tox-docs http://logs.openstack.org/44/655544/2/check/openstack-tox-docs/d650767/html/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/44/655544/2/check/build-openstack-releasenotes/b14bdbd/html/ : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ef39067a02d942f2283ae9a7ee0dd82589030d17","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 16:36:39.000000000","message":"Patch Set 2:\n\nSee comment in https://review.opendev.org/#/c/655526/2","accounts_in_message":[],"_revision_number":2},{"id":"ec619f3724ecbf35d0fcaf8077ec91723137743a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-26 17:34:55.000000000","message":"Patch Set 2: Code-Review-1\n\nplease review feedback on https://review.opendev.org/#/c/655559/1","accounts_in_message":[],"_revision_number":2},{"id":"d5174e9c4f228589a5933dd78fb9ed3cb4a54bcb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 19:46:05.000000000","message":"Abandoned\n\nno follow up patch and adds an unnecessary variable which can just be used by an override","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1375094a4b595ad142564a962c4bfed3e8da0d0e","revisions":{"071f6c6b7eb10f881b0841756c66d85db6c3d996":{"kind":"REWORK","_number":1,"created":"2019-04-24 19:58:14.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/44/655544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/44/655544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/1"}}},"commit":{"parents":[{"commit":"fd7c6f412c52da0c6f6a259396e9f83b48f19a83","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/fd7c6f412c52da0c6f6a259396e9f83b48f19a83"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 19:57:05.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 19:58:13.000000000","tz":-300},"subject":"Adding gnocchi_service_timeout","message":"Adding gnocchi_service_timeout\n\nThe override `gnocchi_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nChange-Id: I2c952b52c2a812df1b7db96abbeb26092a6db80d\nCloses-Bug: 1826250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/071f6c6b7eb10f881b0841756c66d85db6c3d996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/071f6c6b7eb10f881b0841756c66d85db6c3d996"}]},"branch":"refs/heads/master"},"1375094a4b595ad142564a962c4bfed3e8da0d0e":{"kind":"REWORK","_number":2,"created":"2019-04-24 20:12:20.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/44/655544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/44/655544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/655544/2"}}},"commit":{"parents":[{"commit":"fd7c6f412c52da0c6f6a259396e9f83b48f19a83","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/fd7c6f412c52da0c6f6a259396e9f83b48f19a83"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 19:57:05.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 20:12:16.000000000","tz":-300},"subject":"Adding gnocchi_service_timeout","message":"Adding gnocchi_service_timeout\n\nThe override `gnocchi_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nChange-Id: I2c952b52c2a812df1b7db96abbeb26092a6db80d\nCloses-Bug: 1826250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1375094a4b595ad142564a962c4bfed3e8da0d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1375094a4b595ad142564a962c4bfed3e8da0d0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
