)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~660078","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Icbe93a7c35d59cff18f549c0e193beaf7f879cc2","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"Icbe93a7c35d59cff18f549c0e193beaf7f879cc2","subject":"Clarify supported SUSE versions","status":"MERGED","created":"2019-05-20 10:19:20.000000000","updated":"2019-05-27 08:57:34.000000000","submitted":"2019-05-27 08:57:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660078-1558947454271-a45af9bf","meta_rev_id":"edb9026422e6a17cadfce29ee6f99fdcbec54927","_number":660078,"virtual_id_number":660078,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-27 08:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2019-05-20 14:14:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-05-20 12:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2019-05-27 07:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 13:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2019-05-27 07:14:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 12:38:56.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-05-20 13:36:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-20 14:14:40.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-27 07:14:53.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-27 08:57:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"666aa5e6d1bb03007d065fc5afe19355ac512ccf","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-20 10:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd15a799a08ca087991dd82ad8222d29d4cb088a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 12:19:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/660078/1/check/openstack-ansible-linters/53ece17/ : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/660078/1/check/openstack-ansible-deploy-aio_metal-debian-stable/7837f6d/ : SUCCESS in 1h 15m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/660078/1/check/openstack-ansible-deploy-aio_metal-centos-7/f8f33d1/ : SUCCESS in 1h 58m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/78/660078/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/ce86f95/ : SUCCESS in 1h 23m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/660078/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9d0bcbe/ : SUCCESS in 1h 05m 59s\n- openstack-tox-docs http://logs.openstack.org/78/660078/1/check/openstack-tox-docs/4eb4b59/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f4b238ef12ecb594720ffc04684970dbe94f5117","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-20 12:38:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ae28eb1987d839e3a6b6abdd0507c8aaabd4f928","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 13:36:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cd966a55d26825120ac11d4fd7728c8a3cd0676","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-20 14:14:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"08bedcca908171fe8e917b6c957f5490f04f0363","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 07:14:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fce7724752e93524459675d81f9b15eb139d3616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 07:15:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd3bc93bda4e7ff548118b35ed4921547bce0af9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 08:57:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9013c5717cbf23c58edf5a7908d94def137d3075","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 08:57:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/660078/1/gate/openstack-ansible-linters/9406723/ : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/660078/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/2f88b6f/ : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/660078/1/gate/openstack-ansible-deploy-aio_metal-centos-7/7e09d89/ : SUCCESS in 1h 32m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/78/660078/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/399d39e/ : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/660078/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/302cd7b/ : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/78/660078/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/502481d/ : SUCCESS in 1h 36m 00s\n- openstack-tox-docs http://logs.openstack.org/78/660078/1/gate/openstack-tox-docs/b0db976/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36a3f9b89c54452ae3600d43940a2d6837340751","revisions":{"36a3f9b89c54452ae3600d43940a2d6837340751":{"kind":"REWORK","_number":1,"created":"2019-05-20 10:19:20.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/78/660078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/78/660078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/78/660078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/78/660078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/78/660078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/78/660078/1"}}},"commit":{"parents":[{"commit":"8cc12f3fa3969139f9e9eb8b3995ec707c389a21","subject":"Use the dynamic aio scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/8cc12f3fa3969139f9e9eb8b3995ec707c389a21"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-20 10:18:29.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-20 10:19:01.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: Icbe93a7c35d59cff18f549c0e193beaf7f879cc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/36a3f9b89c54452ae3600d43940a2d6837340751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/36a3f9b89c54452ae3600d43940a2d6837340751"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
