)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~663359","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2Frocky~Ic889b94c07d1479da8f7004fce21786e9fb77826","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/rocky","topic":"bp/python-build-simplification-stable/rocky","hashtags":[],"change_id":"Ic889b94c07d1479da8f7004fce21786e9fb77826","subject":"Fix ansible_python_interpreter for identity setup","status":"MERGED","created":"2019-06-05 15:24:25.000000000","updated":"2019-06-07 10:19:12.000000000","submitted":"2019-06-07 10:19:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663359-1559902752137-94b57270","meta_rev_id":"a515e33d6cb5e82fe719b5f9f609053084c5fce0","_number":663359,"virtual_id_number":663359,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-07 10:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-06-06 15:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-06 19:48:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-06 19:48:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 15:15:48.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-06 19:48:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-07 10:19:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06d9bde6e5c23803921eba449d2568a39dda7653","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-05 15:24:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"33ced71f2cc081d7e00cfbf98dbd8eb0922b341d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 16:29:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/59/663359/1/check/openstack-ansible-linters/1a5a9fe/ : SUCCESS in 6m 06s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-centos-7/2b543c8/ : SUCCESS in 32m 29s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-opensuse-423/91aeea4/ : FAILURE in 8m 55s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-opensuse-150/92ecdbe/ : FAILURE in 10m 55s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-ubuntu-bionic/681a7b8/ : SUCCESS in 27m 26s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-ubuntu-xenial/d6441ad/ : SUCCESS in 24m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/663359/1/check/build-openstack-sphinx-docs/2ad31bc/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0f33ac401cf1178c45f712c9736864026abab57e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-06 15:15:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26562b7150c1607bd150e453a425d63c2b9cee4d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-06 19:48:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e49f9d40157af3aaef87d8ad20922e5097721a99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 19:49:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5aa16fea2160864a740957b746d704fcbb89c880","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 20:33:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/59/663359/1/gate/openstack-ansible-linters/2015679/ : SUCCESS in 5m 39s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-centos-7/c7baa0e/ : SUCCESS in 30m 28s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-opensuse-423/a1261f1/ : FAILURE in 8m 11s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-ubuntu-bionic/c28dadf/ : FAILURE in 15m 49s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-ubuntu-xenial/85ade27/ : FAILURE in 15m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/663359/1/gate/build-openstack-sphinx-docs/dfef8ec/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e26c3a288b356b52568df2ad343bec05449b2c9f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-07 08:43:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60653296e54ca4c4259cb625caea48db2952446a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:43:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/59/663359/1/check/openstack-ansible-linters/28fa1de/ : SUCCESS in 6m 11s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-centos-7/f5bc634/ : SUCCESS in 35m 45s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-opensuse-423/f80a31b/ : FAILURE in 7m 49s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-opensuse-150/9266fb9/ : FAILURE in 8m 34s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-ubuntu-bionic/80f15fc/ : SUCCESS in 32m 46s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/59/663359/1/check/openstack-ansible-functional-ubuntu-xenial/7940e23/ : SUCCESS in 51m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/663359/1/check/build-openstack-sphinx-docs/a962961/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c8f1f6c41dfeb10b4ed86b99b4d28248a43aa3fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:43:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5541f8de5577020f442e6557f926a62e68f4f5a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 10:19:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/59/663359/1/gate/openstack-ansible-linters/eaa82c7/ : SUCCESS in 5m 17s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-centos-7/0893f85/ : SUCCESS in 30m 12s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-opensuse-423/866c485/ : FAILURE in 8m 44s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-ubuntu-bionic/62124fc/ : SUCCESS in 26m 33s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/59/663359/1/gate/openstack-ansible-functional-ubuntu-xenial/5aee7b0/ : SUCCESS in 25m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/663359/1/gate/build-openstack-sphinx-docs/a23e729/html/ : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0fe09ddef17b203916a29dff9271cc4cc8023611","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 10:19:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f72585873c6c7b6bc1581ec9465c6c12a47b6425","revisions":{"f72585873c6c7b6bc1581ec9465c6c12a47b6425":{"kind":"REWORK","_number":1,"created":"2019-06-05 15:24:25.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/59/663359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/59/663359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/663359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/663359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/663359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/663359/1"}}},"commit":{"parents":[{"commit":"2c4672941312b421bc9a7e08a41d73ae8a936470","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2c4672941312b421bc9a7e08a41d73ae8a936470"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-22 15:35:02.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-06-05 15:24:25.000000000","tz":0},"subject":"Fix ansible_python_interpreter for identity setup","message":"Fix ansible_python_interpreter for identity setup\n\nI04e853fb582d7b39708a2fb6ed854b4e458f06d9 fixed this for\ngnocchi_service_setup, but did not fix it in gnocchi_identity_setup.\n\nChange-Id: Ic889b94c07d1479da8f7004fce21786e9fb77826\n(cherry picked from commit e0f77278923ba6af8d0760656717d8d8dfabe81c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f72585873c6c7b6bc1581ec9465c6c12a47b6425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f72585873c6c7b6bc1581ec9465c6c12a47b6425"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
