)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~664089","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2Fqueens~If79f08a57ab519dfec977697ad7225e921b38b93","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"If79f08a57ab519dfec977697ad7225e921b38b93","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:24:28.000000000","updated":"2019-06-11 22:51:41.000000000","submitted":"2019-06-11 22:51:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664089-1560293501913-d9eb55b5","meta_rev_id":"f1cae8d97c24f30c90f9faf38de48ff9f74482e8","_number":664089,"virtual_id_number":664089,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-11 22:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 21:55:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 21:55:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 21:55:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 22:51:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"088799db4ad0027cbaf15e85bfd20ae0f8b7732f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1cc2df005fd6f2e5431ce9436c780e86242ba1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:29:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/664089/1/check/build-openstack-sphinx-docs/91da50a/html/ : SUCCESS in 3m 49s\n- openstack-ansible-linters http://logs.openstack.org/89/664089/1/check/openstack-ansible-linters/3f63023/ : SUCCESS in 6m 18s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/89/664089/1/check/openstack-ansible-functional-centos-7/fee104e/ : SUCCESS in 51m 14s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/89/664089/1/check/openstack-ansible-functional-opensuse-423/3d0aa5b/ : FAILURE in 8m 44s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/89/664089/1/check/openstack-ansible-functional-ubuntu-xenial/4b54606/ : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0458435aa7ed5ec63ffd6a618feeee15d9c98bd5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:53:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e812a9e856cb95d4445516c6e13297e826c2fd43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:43:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/664089/1/check/build-openstack-sphinx-docs/ee04eb6/html/ : SUCCESS in 3m 16s\n- openstack-ansible-linters http://logs.openstack.org/89/664089/1/check/openstack-ansible-linters/1ecafc8/ : SUCCESS in 6m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/89/664089/1/check/openstack-ansible-functional-centos-7/b170d1d/ : SUCCESS in 43m 10s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/89/664089/1/check/openstack-ansible-functional-ubuntu-xenial/8d37071/ : SUCCESS in 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7260e82b3e6167cad483fe3bd826ad04e0f730cb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 21:55:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ac5907f2d99533a07d936930e5fa29e273dada8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:55:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7af5e9f71b17a3c2f3614d3707edc7844a49729","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:51:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/89/664089/1/gate/build-openstack-sphinx-docs/5c52637/html/ : SUCCESS in 3m 01s\n- openstack-ansible-linters http://logs.openstack.org/89/664089/1/gate/openstack-ansible-linters/ac7a91d/ : SUCCESS in 5m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/89/664089/1/gate/openstack-ansible-functional-centos-7/644429b/ : SUCCESS in 47m 15s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/89/664089/1/gate/openstack-ansible-functional-ubuntu-xenial/1b9b43f/ : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2dc6236abd89764ee2b64e6d34585a86d5e73aee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:51:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51b558103ef9c90b19d1aea455ca3a121eb5dcf8","revisions":{"51b558103ef9c90b19d1aea455ca3a121eb5dcf8":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:24:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/664089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/89/664089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/664089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/664089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/664089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/664089/1"}}},"commit":{"parents":[{"commit":"60ba9416cba04711b016ee41f3ffb7fded33ea9c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/60ba9416cba04711b016ee41f3ffb7fded33ea9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:24:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:24:24.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: If79f08a57ab519dfec977697ad7225e921b38b93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/51b558103ef9c90b19d1aea455ca3a121eb5dcf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/51b558103ef9c90b19d1aea455ca3a121eb5dcf8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
